)]}'
{"id":"openstack%2Frequirements~715641","triplet_id":"openstack%2Frequirements~master~Ia1aad0c61ae9959b3dfa4e148f0a3ce332d774af","project":"openstack/requirements","branch":"master","topic":"test-docutils-0.16","hashtags":[],"change_id":"Ia1aad0c61ae9959b3dfa4e148f0a3ce332d774af","subject":"DNM: test docutils-0.16","status":"ABANDONED","created":"2020-03-28 17:30:45.000000000","updated":"2021-03-07 01:15:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe2400ab54f05bd7f4e34bbb362413de69873787","_number":715641,"virtual_id_number":715641,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-14 18:31:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-08-14 17:03:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 18:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bcb21d8f399a803d29573d4fff6c7587b4141ac9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-28 17:30:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05400f29d7e30ea4cdbdede5088590dece8f3948","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-28 17:34:29.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e2a23714c53535c228a827a3ed210a45bcb87c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 18:58:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a521947217e4fda9800937ef10f4166 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7413ea1518ea4d719bca454dd50d1235 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba479d07593544068028232f41cf6ada : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72706db25e864cfaacfacbb715228b0a : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a13360a600b45d1aad3da8cc648738c : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e7e09424bac46d5bf2a2c037b45b6c9 : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/910c4bbede6c40f39ae6dd750a62cb14 : SUCCESS in 2m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/029fc1f2b8a14a9e8105bd7d9afba664 : FAILURE in 7m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/67492f632a5e4494a80b2c2746dcba76 : SUCCESS in 7m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4651863fc6cc41e3a246eece41ab0d2d : SUCCESS in 4m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5df3194b1cf34c53aa5140211d1693a3 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf2ed1c7992f4ba6afa178c785fd9d09 : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/102aca80d3524e24861e86ff95b0995a : SUCCESS in 18m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2f6c471380c94f9b8264e39b96541bd6 : SUCCESS in 29m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/494d74d7cb074a17864fbfb063720a4c : SUCCESS in 14m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3adc66bb6bbc4b6ea940e54ce2adf1e3 : SUCCESS in 12m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8babfebc1ccb4563b95057500b8973bf : SUCCESS in 19m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9806a913a3084e1cb0b07648dde652d4 : SUCCESS in 4m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/86f260bf54564a859df3fc2fead40249 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3208c5f5070400db35936549c6f12e1 : SUCCESS in 5m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/21119ac32c8d4eb88cd2911a969ce8ac : SUCCESS in 11m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8ef062d4b6d84cd89685e16694eba60c : SUCCESS in 16m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5931019f9cf94ff19324ca17dab9700c : SUCCESS in 57m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e052bcea617847ba8ca9ad42f929628c : SUCCESS in 1h 09m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56964599d45648fb926125f79dc456a6 : SUCCESS in 1h 25m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5c2e7f71722a45a0a439fd3b852e6e06 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8b1dfb6e0d9d8ac3e92ca1166ef2ee30c2c859be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-14 17:03:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd389c8baf556470d69db5a28188870530959170","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-14 17:03:39.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"e5bcac750ff71bcda2ef6de8bb9735e95ba3efa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 18:12:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/deb56e1a606a4bea9609633848bef932 : FAILURE in 1h 05m 19s","accounts_in_message":[],"_revision_number":2},{"id":"66c3ad083b3431315d3e7d36545f33dcd91dd881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 18:31:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e6410a404f44e82915542b957cffa52 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bacc56746b174f7cb202d021c909639c : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b54ff9bfbfe4929b06bd6b698c2043c : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/913849e806dd4fa980a5abc842f088ef : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a6ce6c1420d4f6bafcde3213271a050 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24c5a03991f1426187460ce79b772ea8 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ed39bbb1e16549308a6114dd7ad9dc4b : FAILURE in 6m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2a20fa9ca6f140289ad9ed3f67e2005f : FAILURE in 12m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0e7af24f4e22425480faaa980d395a03 : SUCCESS in 8m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3907ba9b6fb340269b0d86b660d22180 : SUCCESS in 4m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8692237a2126404498cbf6e1b20dd98f : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f20735306c64345b548b194db9d1647 : SUCCESS in 9m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cc4bd71d236f475bb7814c088d116a2b : SUCCESS in 12m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/acfddfe0e1ff4954ae9da7b3bff2675f : SUCCESS in 12m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0ebdc9f375524f1eb3f3e4d332588208 : SUCCESS in 22m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0543d4b4d0314bf3935ed2d980c310b7 : SUCCESS in 12m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9bb1a803135641b0970ffdfbd59ec195 : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3d9e8a3b6884ae58e3ca5b2f45c87a0 : SUCCESS in 16m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/04483652bc5c4682af37f741d045ab36 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/28faa2672d4f4028afd281992987a35c : SUCCESS in 6m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/903dbd523fdd4c3ea8697c58a267cfb2 : SUCCESS in 9m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3e23be7a27c46c4a13062c0dbda366d : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f34b863897a6420c9047c95cfc30a635 : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":2},{"id":"fe2400ab54f05bd7f4e34bbb362413de69873787","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-07 01:15:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83af27157323366875f30a24fbe9865f0edecdee","revisions":{"ea1b44ccaaa0d8b1c78519809f77849bfc7b2108":{"kind":"REWORK","_number":1,"created":"2020-03-28 17:30:45.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/41/715641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/715641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/715641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/715641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/715641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/715641/1"}}},"commit":{"parents":[{"commit":"0b01b7f94bd23cbf3f23d9dc5ad3bed48a6835e2","subject":"Merge \"update constraint for python-openstackclient to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b01b7f94bd23cbf3f23d9dc5ad3bed48a6835e2"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-03-28 17:30:23.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-03-28 17:30:23.000000000","tz":-300},"subject":"DNM: test docutils-0.16","message":"DNM: test docutils-0.16\n\nChange-Id: Ia1aad0c61ae9959b3dfa4e148f0a3ce332d774af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea1b44ccaaa0d8b1c78519809f77849bfc7b2108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea1b44ccaaa0d8b1c78519809f77849bfc7b2108"}]},"branch":"refs/heads/master"},"83af27157323366875f30a24fbe9865f0edecdee":{"kind":"REWORK","_number":2,"created":"2020-08-14 17:03:12.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/41/715641/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/715641/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/715641/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/715641/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/715641/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/715641/2"}}},"commit":{"parents":[{"commit":"2ca5511020f07bb7107afaa42e9d917a4e4c1370","subject":"Merge \"update constraint for os-brick to new release 3.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ca5511020f07bb7107afaa42e9d917a4e4c1370"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-03-28 17:30:23.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-14 17:03:05.000000000","tz":-300},"subject":"DNM: test docutils-0.16","message":"DNM: test docutils-0.16\n\nChange-Id: Ia1aad0c61ae9959b3dfa4e148f0a3ce332d774af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83af27157323366875f30a24fbe9865f0edecdee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83af27157323366875f30a24fbe9865f0edecdee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
