)]}'
{"id":"openstack%2Frequirements~715876","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I0f92792f1f28861582c355808db28ec4f91df911","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"I0f92792f1f28861582c355808db28ec4f91df911","subject":"Bump subunit to 1.4.0 in train too","status":"MERGED","created":"2020-03-30 08:29:48.000000000","updated":"2020-03-31 00:19:49.000000000","submitted":"2020-03-31 00:19:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715876-1585613989794-ea78e57e","meta_rev_id":"cd7266ae711567b48e32a9826fefdfc1aaf8aac2","_number":715876,"virtual_id_number":715876,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"date":"2020-03-30 19:28:18.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 00:19:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-30 14:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-30 15:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-30 15:04:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:21:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2020-03-30 14:42:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-30 15:04:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-31 00:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dcc7d726bffce4c8101259e47c45f416c4427cfd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-30 08:29:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e3aa4071332e026c33bfe46d61177e37ddcb00b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 10:09:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e7bba51b995541149561b04bb6f13a6c : FAILURE in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea7690820d84f969f457d3762b17127 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94820bc3e56d484483fc095eb47e660c : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd00df3ada0a4297a4604532b33a3e17 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52efb22405924eac9b29bb8eeb6789af : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d787834ff244d3ba9483b5d051a1df2 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d2c9fe6be814dc0864ae4aaec527595 : SUCCESS in 3m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e2a21bc82d984a3eb20a1789671d6e7c : SUCCESS in 14m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1ec6d7ff4bf446d99d14813576f69060 : SUCCESS in 7m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0e0ec5a304b44ba180bb97f78d592449 : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7851b72df6d24f03a9203fba0d3b1e38 : SUCCESS in 9m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f9c28ecf8bc942f695067b6a093e2211 : SUCCESS in 6m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/47b4f68e544243e2aebbb3abb2c52ac8 : SUCCESS in 6m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b51d193c623b4d60a1ac5941655d2085 : SUCCESS in 8m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9404063cd9cf404cb7129bf425d62dd3 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8750453c3272463f99b728a5af94ad7d : SUCCESS in 9m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7e87ff11646c47b8928ecfba2b33a4b0 : SUCCESS in 20m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8e4189fc5f9d48258425f3e8f14f16d0 : SUCCESS in 22m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3b07bd300e124d6987524a1e4eade1d8 : SUCCESS in 24m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2cd1492dc66945698dc2a53fc499bbad : SUCCESS in 22m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1139c28e89df4d6d95ad5343420538c2 : SUCCESS in 12m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1f515d20c3b0468fba46449570d38f04 : SUCCESS in 11m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4491a5ce6d8f42edb78731ac0459e5fb : SUCCESS in 18m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/444a87b0109c44998071c7db8993fae5 : SUCCESS in 21m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7646fe4d66204ec7a4c3836241d65644 : SUCCESS in 5m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9fa07394539444209b7eb42eb49619cc : SUCCESS in 9m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ff5289f0439a44748ef4bc5e369ea0b4 : SUCCESS in 9m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e8576315e4040818ecb82ef551d90cb : SUCCESS in 1h 14m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/81211b9e2a1a4d2691601d93cc18e56e : SUCCESS in 1h 29m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cb83defe61c44c33845940acb43dd732 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":1},{"id":"188b77f2e593fd5277531cba0607c822480c0a5d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-03-30 13:21:44.000000000","message":"Patch Set 1: Code-Review+1\n\nI can confirm this will be useful for Glance.","accounts_in_message":[],"_revision_number":1},{"id":"1932fcc27e79b201cfa9a8a9795a9cac4fc43606","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-30 13:29:15.000000000","message":"Patch Set 1: Code-Review+2\n\nNormally I wouldn\u0027t want to risk raising something like this in a stable branch, but since it was something that needed to be addressed with carrying patches and appears to be mostly safe, I think it may be worth the small risk in this case.","accounts_in_message":[],"_revision_number":1},{"id":"081b875dad53d1ce969a87ab930522baad844bc4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-30 13:51:10.000000000","message":"Patch Set 1: Code-Review-1\n\nit\u0027s understandable that downstreams carry patches (part of the point of using a downstream).  I\u0027m trying to determin how this helps openstack and why it should be allowed within the stable policy but don\u0027t see a reason to update subunit to 1.4.0","accounts_in_message":[],"_revision_number":1},{"id":"b0992b028307dda724b9614ca8f0935944805ef1","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-30 14:04:41.000000000","message":"Patch Set 1:\n\n\u003e it\u0027s understandable that downstreams carry patches (part of the\n \u003e point of using a downstream).  I\u0027m trying to determin how this\n \u003e helps openstack and why it should be allowed within the stable\n \u003e policy but don\u0027t see a reason to update subunit to 1.4.0\n\nIf I understand what glance people told me correctly, one of the relevant patch in 1.4.0 fixes an crash while running subunit filters against the glance unit tests.\nOther patches are relevant to py3 support, which is a must in train.\n\nSo please reconsider.","accounts_in_message":[],"_revision_number":1},{"id":"55bd072fda1a7d21d9a687794d18543c11c6898c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-30 14:20:48.000000000","message":"Patch Set 1:\n\nIf it fixes an openstack bug that\u0027s a reason to merge, mind linking to it in the commit message?","accounts_in_message":[],"_revision_number":1},{"id":"4aef89d17f2f2b6db742303cfc1bfaf78fcaca71","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-30 14:42:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"19c11e2599b1050da78477fc63fbf5573115de99","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-30 14:42:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"10a275cddf527de57d656298f339200c310318dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-03-30 15:04:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cddb0984c1199c3702bb35f45b4626937633ffc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 17:34:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7684f2d123ac49faa014a4f99f85deb0 : FAILURE in 6m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d2abfaa3451414dbb9e7cd200545729 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e7d57f6c2a9457fb73d32b3bd0c3b30 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5b8b867e4f74314bf44e4479c4654b5 : SUCCESS in 9m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6abeb129999940c4922ec9c0b68054fd : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f948685dff804e02bdcc2cdbdf033f0d : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af3f9f45b68341319ac790aba5c2340a : SUCCESS in 5m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/db64e95a445d470586eb4c20c73fe5bd : SUCCESS in 8m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0aba50c097814a48bee53b0936214d17 : SUCCESS in 11m 45s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/92b04ef4cbe7469b9bb0e2206ba2f4e7 : SUCCESS in 12m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b50c428d82f349d3a4f10189b592999a : SUCCESS in 11m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/780590eb9c2c4829993e64d708717f1c : SUCCESS in 11m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b6f9a9650b9f4a83881b0a8f2597aa38 : SUCCESS in 12m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/86d4dd499a994d3887199c5bba84ace0 : SUCCESS in 12m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ea82d6ba1a344b1e910fad195b0d0b2d : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ef8e57a8bc44dbca6db97734ac8f818 : SUCCESS in 7m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5d76694c38b341469b486363064de86b : SUCCESS in 22m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b4d81d776da14e5bb3d175f744034e46 : SUCCESS in 22m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ed70eb1764434c68ad5456d04e6245aa : SUCCESS in 23m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3a5c2d9f92e431ab05202490ea93cee : SUCCESS in 25m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/51ef629777654e7dad1c32c3bae560e6 : SUCCESS in 13m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a9030564e1c845a1be16569c14e8e8d1 : SUCCESS in 15m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cc36ca9438d34848a4724083e49a70b9 : SUCCESS in 16m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d42dcf9556f247afa4e2b903a1585225 : SUCCESS in 19m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0960b4e8271c49738416889cf8959964 : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4291c4ec96694286b951ff21a49cddc2 : SUCCESS in 11m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74b2616e71bd4802ae0cc93a6b6b010f : SUCCESS in 10m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/952e9fbfcf3542538d1f8dcf9ec82912 : SUCCESS in 1h 11m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f84746ad27de4c279677dd128960d4bf : SUCCESS in 1h 34m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bfe2e7aa02b44efeb81767d91cf91b16 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":2},{"id":"198b1a26d4b5efad8d2e1cbd6f9d4884b912cd23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 17:34:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5aafb293704dadb5406b3ec1059a0dc34923bed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 19:21:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363dfc04a7834cbda7528cef53eef1f7 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e78d9098ca94c00b2afd3e9d2534827 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c946c9b58b3c4493959adfe9138c4c2e : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68a871eb5bb34c409bad544af56fa12e : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b198594adfad4dcd926f4e203fb4b8e5 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e77940fcf41b40d18236c574761c5dcf : SUCCESS in 5m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/96f16ca948a04a7a84130d7ecf2dd712 : SUCCESS in 8m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/639b23612699463f86c55ea90e25ba5a : SUCCESS in 7m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bec518eb93f64f31b5643eaa7475ebbc : SUCCESS in 13m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f3993d746c4546fe831b4bf26f562493 : SUCCESS in 14m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4e1bf97754234d6ba1efaed5038a53b1 : SUCCESS in 5m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d19c67827cea4ce49d5efcb1bec49d1c : SUCCESS in 6m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d35ba5f5a6184ddba4400510a405ac5c : SUCCESS in 12m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4fddaf2c92f64c3d9021218811f7a4d3 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a951a6dfd8304793ab03fc6d9db645cb : SUCCESS in 9m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/14832a9806a841939900ae2dad6b2327 : SUCCESS in 15m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e1339e48fad1425bac614735af8b3603 : SUCCESS in 25m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/40c336063541420db5db89d3f81faa7b : SUCCESS in 22m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/476fe64b03af4b2ab38972ac9c672516 : SUCCESS in 20m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8ae4f2a04c9040f881139b172ed8b6ae : SUCCESS in 13m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/996d4471360b443782d3d5d387071045 : SUCCESS in 16m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1cf88a391f19439ab316f5b3fc9c56a0 : SUCCESS in 16m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/194d4b6ca8014c48a713a32a2f8bb966 : SUCCESS in 19m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/944bcc6dd7c5471687a270b032643f40 : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7cf4460827c94da097d3b581d5ca17df : SUCCESS in 9m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4c783b67b17e41e6b31eb6a760de0699 : SUCCESS in 12m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fe6cfacfccd4e4e8ee1a3533130f7b1 : FAILURE in 14m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d5f096a49cb7417ba70e6bce1a372777 : SUCCESS in 1h 37m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c145a08c61684666bb373c7ff3dd6c66 : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1c2e79809b8f0e2bfe8abe65c08b5b46e5b9a978","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-30 19:27:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b1e0603c01a7f384b6cceda19db5be5c58ef5f01","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-30 19:28:18.000000000","message":"Patch Set 2:\n\nInfra error in neutron-grenade-multinode:\n \"Failed to connect to the host via ssh: ssh: connect to host 10.209.64.26 port 22: Connection timed out\",","accounts_in_message":[],"_revision_number":2},{"id":"2b3f60154798bf5e881946af9bcde082da0d8a86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 21:55:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d9535710b6644516b339e568260d6cd6 : FAILURE in 2m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7547faa696a42dd96caf38303771538 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e16b6ef1109c4d02b3c912dac841f5e1 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e83f8cf6b7114c15a4daaae1b1887286 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbc16fdb97cd4065aacae1538d80587a : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8459404817194f19bf889f9a4c9c8b8a : SUCCESS in 6m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca8a82b6298743948bf5ccf8a14f3b36 : SUCCESS in 4m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fcfaa1ed5dc546338330e14a77a7f0b0 : SUCCESS in 9m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce1d8492f412491caf0308f8c671a450 : SUCCESS in 9m 44s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0b80e73972f54e41802b5ee5377c2be7 : SUCCESS in 9m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cde5755b2d9448ca9625ff66bf10feda : SUCCESS in 10m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/65e386800679431eae5b993ba601a94b : SUCCESS in 6m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a0d24169fe7e4e1a98fc1f3949672a85 : SUCCESS in 6m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2c04026f6cc84bb99b5063cf9032051b : SUCCESS in 12m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/06cabcac2fb04b78adcb70552631a178 : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d15efa55a88f4395a0ff9153fece36a9 : SUCCESS in 8m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/10d7b8554a7d4c7ebce2824d543a650a : SUCCESS in 24m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/420ca40010a04cf5be6c298ed64fff21 : SUCCESS in 20m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/eb8b218d60954d169c3220864f40445a : SUCCESS in 22m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c940c389ff704f53a68cb6cb4a2bdddb : SUCCESS in 21m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b35f808b18d644faab42538eea18a119 : SUCCESS in 15m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/49bbc4bdb393485892c2ba73da379c45 : SUCCESS in 14m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/93d3bfb9f41d4ab897d48a409f15306d : SUCCESS in 16m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42f13d1f09544e2f9122d5d6b76742d6 : SUCCESS in 19m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a879d31a7164389b8aa126ff21c27f5 : SUCCESS in 5m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2ef92db0c4414d75bc7f2c5d6dc1aa01 : SUCCESS in 10m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b0b5d876df224e78803baf60ce517914 : SUCCESS in 8m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b18d06053ad48c689c0a3391e165d68 : SUCCESS in 1h 08m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3ce075bf3f9740718079f199fbc9a54b : SUCCESS in 1h 47m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3ec341d835f34aa0a6f24a6844eb54d4 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":2},{"id":"d934390e1843134dc676a72b9b96397e78bad459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 21:56:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f450af21e7a792df9fe3b248629ddd1b83bb984","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 00:19:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cd7266ae711567b48e32a9826fefdfc1aaf8aac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 00:19:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20a3d545b1c8410eb57ca4afb6a1c3eb : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88a67f087a5743209188cdad1c256f4f : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e94983feae8f4f198cd2ff342da354a6 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3108d5cb54894fd29f956280b20cfc9b : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cb67ad1fd0e4b7cb3f2c2a8d7cd27f3 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dfbea5e5903640e59e1f6365375e9499 : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c5c30964bb174e43853ed16d242eab95 : SUCCESS in 10m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c95c41c488384a1d86155f75a2d0484e : SUCCESS in 11m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/33de9905d14b49d697f4c87477e45480 : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/972ee72f1dd54089b2f775728dc936a7 : SUCCESS in 9m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1c02e0a0a3bb4d43af185d5938571ebe : SUCCESS in 8m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ac6a1b25b4984d388eb70da86b0eee27 : SUCCESS in 6m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/34c9b732b5d8466e9254203269d829c0 : SUCCESS in 14m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3cea1fd52e714a66bf1926163d3772c8 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6784e8b6b43949868ece2a1bcfd61864 : SUCCESS in 22m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ebba947e33ca4ad0a371da01a0980424 : SUCCESS in 14m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef776fe8a6594153aea146a055debcf2 : SUCCESS in 34m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7999495445194719b817ce6afc99eb23 : SUCCESS in 25m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/15c8ef2d6a094c13aa675e8b766acbbf : SUCCESS in 30m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/05aa1c1348ac49bca00e68ef5c265a72 : SUCCESS in 27m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/941350d3cae548c9a4eec7512a3bfbab : SUCCESS in 14m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/acc884f11e6d4b4eb9e8b74ee3c163c5 : SUCCESS in 16m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5cd4a34dfa804c3b88dfc17ce00e50ff : SUCCESS in 28m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a435f2d0d2c44bbfa86a6b75d2c6d96e : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5a3db3d4ee67447598e76a0ae1e489fb : SUCCESS in 9m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d8e1b9975d134b95bf4f711726195266 : SUCCESS in 12m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/507de6390a1146279e5188cca93e43b4 : SUCCESS in 1h 10m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c28e551988546a3aff2c6d95ac31d4e : SUCCESS in 1h 42m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1de306a9e77448ffbfd530c193aac89e : SUCCESS in 57m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9e0766c87eb4da7ff3443a5f5aa22b82bf22220d","revisions":{"d1c656c1693e7cc26af9af33550b14d5b55b4c76":{"kind":"REWORK","_number":1,"created":"2020-03-30 08:29:48.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/76/715876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/715876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/715876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/715876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/715876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/715876/1"}}},"commit":{"parents":[{"commit":"01bd6d70a8fbd569b3645ce58be2ce1530ec5353","subject":"Add Pygments to upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01bd6d70a8fbd569b3645ce58be2ce1530ec5353"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-30 08:27:29.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-30 08:27:29.000000000","tz":120},"subject":"Bump subunit to 1.4.0 in train too","message":"Bump subunit to 1.4.0 in train too\n\nThe recent subunit 1.4.0 release includes several Python3-related\nfixes that can be useful to have.\nAt least one downstream consumer (Fedora and then RDO) had to carry\nvarious patches from to make subunit 1.3 work with Python 3:\n\nhttps://src.fedoraproject.org/rpms/subunit/c/91fb7acb47242d51f9a222152a0b30b8f9b4e95c?branch\u003dmaster\n\nChange-Id: I0f92792f1f28861582c355808db28ec4f91df911\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1c656c1693e7cc26af9af33550b14d5b55b4c76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1c656c1693e7cc26af9af33550b14d5b55b4c76"}]},"branch":"refs/heads/stable/train"},"9e0766c87eb4da7ff3443a5f5aa22b82bf22220d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-30 14:42:00.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/76/715876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/715876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/715876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/715876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/715876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/715876/2"}}},"commit":{"parents":[{"commit":"01bd6d70a8fbd569b3645ce58be2ce1530ec5353","subject":"Add Pygments to upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01bd6d70a8fbd569b3645ce58be2ce1530ec5353"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-30 08:27:29.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-30 14:40:14.000000000","tz":120},"subject":"Bump subunit to 1.4.0 in train too","message":"Bump subunit to 1.4.0 in train too\n\nThe recent subunit 1.4.0 release includes:\n- several Python3-related fixes\n- a proper handling of \"extensive logging\" from unit tests\n  (see LP#1813147)\nthat can be useful to have.\n\nAt least one downstream consumer (Fedora and then RDO) had to carry\nvarious patches from to make subunit 1.3 work with Python 3:\n\nhttps://src.fedoraproject.org/rpms/subunit/c/91fb7acb47242d51f9a222152a0b30b8f9b4e95c?branch\u003dmaster\n\nRelated-Bug: #1813147\nChange-Id: I0f92792f1f28861582c355808db28ec4f91df911\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e0766c87eb4da7ff3443a5f5aa22b82bf22220d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e0766c87eb4da7ff3443a5f5aa22b82bf22220d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
