)]}'
{"id":"openstack%2Frequirements~716491","triplet_id":"openstack%2Frequirements~master~Ibef65d96adc52518f7ea5dc7608daddd4d1b6276","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ibef65d96adc52518f7ea5dc7608daddd4d1b6276","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-04-01 06:18:41.000000000","updated":"2020-04-03 10:51:37.000000000","submitted":"2020-04-03 10:49:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716491-1585910962934-e8a3316c","meta_rev_id":"e2ae858f344f21522b91cd1c2ef68f5194fa2154","_number":716491,"virtual_id_number":716491,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 10:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-03 06:26:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-03 06:26:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 06:26:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-03 10:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef182637255ce4e450fcf0b6492fc687eebb2e22","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-01 06:18:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"279179ef14120832bcbee6e522fa5096ca958396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 08:55:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/552e727873fe462dbe90198500f42a80 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adeb58e5c3a3475aa9105215dd526cbe : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a8237dd2ac4c84a59e178d2bd95d0c : SUCCESS in 6m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd31049684f5452c9135278684238179 : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15b9879d73c24bda90e5b026e012dc54 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0dbbb19ed4c450c9f3146ebb0fa09e0 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd4d251d83204495bfadfe8c76262ecc : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aae22a14e1194a1b8841c51fd8e29a3d : FAILURE in 13m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d7c5104c721045279e90e2643ac88999 : FAILURE in 7m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00739854826d4387af5cdc5d677e37ed : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53598dc328ee41d78edaae0115165a60 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d6d578603d240f985b3795a0ceb4c44 : SUCCESS in 10m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/93632c1d3b424eaca6c22b7c7ef5f5fd : SUCCESS in 13m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/250af4c888624a119931b2bee1fcbb98 : FAILURE in 29m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d9edf0dd4306459ba64b6dea55f4c907 : FAILURE in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/935726c8494d4cab93558229360096a2 : FAILURE in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/008c6438017f490abcbf897cbda90344 : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/884aef0ce05d4aa99d140ff4b3c85afe : FAILURE in 5m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d50cfad08c52465090785dbc6cf34728 : FAILURE in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40fa290095524160a3c6e919d57b196c : SUCCESS in 8m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b3e5758569934e7e9e41e44f627cf613 : SUCCESS in 12m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/17b1cf10608745d1a3a22cac6827120f : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8324b844eadc4d04927077c6436c7ace : SUCCESS in 1h 00m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d46cd37f3e18486ba2b7907830dab82f : FAILURE in 1h 13m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bee35e574fec4216bd4b603c4e9ac6c4 : SUCCESS in 1h 53m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/97795416dc574623bc8467359f636e72 : SUCCESS in 1h 07m 40s\n\nWarning:\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Comments left for invalid file stdin","accounts_in_message":[],"_revision_number":1},{"id":"e142f35b37e9e5802f1ca77b2323b73540334123","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 06:14:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1cad86f1697596b863d528ad5cf1c4e9ea6b78c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 08:35:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8826178777a143c58ea22d22d019d1fd : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1770543591404528bc0ce4eb6bb0f429 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31bc5475d37b4192ae15855ee7ec913d : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dcdb30ad97cf4cd9ac703d157bace690 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a20ffd87f734a6d983283e9a077e0ac : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dee9be59f38a4f4291d872264088c91e : SUCCESS in 6m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9fbaaeafd9b4c86936753d5a68ee787 : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/885f034bddce4dafbf25b292269a8299 : FAILURE in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/40d3a79643cd4447abc19b46b2d2140f : FAILURE in 6m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/25b4d5b3816d47cfa920bc851a694fe1 : SUCCESS in 7m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/45f1ccc2404d45a8a9781619dd39a52c : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea3389cc83b84104ac914838e11894a1 : SUCCESS in 9m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d308ce10aa4c4994bfbf201f7669c535 : SUCCESS in 27m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/462c8787e4b84cfbba0c79ca2ba4e831 : FAILURE in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc0e0a097edd4bd78f50199a7c8228c5 : FAILURE in 15m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0ed98d9b68504fbda37a7d0e60be5390 : FAILURE in 16m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e06dcf8ab1b44397b69eaf48be139c33 : SUCCESS in 16m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0d7acf084bf0495fa848f19b7011677c : FAILURE in 5m 56s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/884c6b1498bf47f7a13f88fba084d9cb : FAILURE in 7m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e26e33d3cd245d7b8d1a2b511c5813c : SUCCESS in 8m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/92191ede838d4a9c833412493f146850 : SUCCESS in 12m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44ed1c99a1964419b402114bd63cae8d : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/434f20d329e74041b76fd43461cdbc07 : SUCCESS in 56m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c18544aa7454864b65909b4170dfafd : FAILURE in 1h 10m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b84ca53a6ea74af5adf83c90ded7bf4f : SUCCESS in 1h 27m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5a61039cd1ac4252b712dbd43dcdb234 : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4351a2a86395ff02b837fc98fc22ce2a7f216e60","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-03 06:22:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9ccf60f774b330a94dbb6c4aeef7aebb5cc17951","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-03 06:26:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa7822fd27161e871ec45fb520a9818873cea0c0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-03 06:26:27.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"3d4755fe8dbf42dc24c58e9261c42aeb4d68bfc0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-03 06:26:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e8c154ec15a7f3aba7711943bfe9db93bbdb16b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 08:57:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53ae29b4f66b47a5a02e543a2081395f : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f960f28f93495cb066d812158c5979 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ea4cc8bf1a747a3bef7cdbf381ab15c : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed4deac31fb74be18480ecbdbab0eaa0 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3cb379d2bc64ddead9ed05926571e4a : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d672725d35cd4513b1b67a2e2fa29a39 : SUCCESS in 6m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c6ff7351d8a4d7799832494a7527d9f : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/22194af010014dfcb9fe44c24d29cc1b : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/46ffe422784c4897b8ae1a79a6874215 : SUCCESS in 10m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9a482b773eae455d859f8f2d566afa53 : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/58ae0eab50bf4a609ca370580fe32950 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9f765ae2968409e9891468a3850e134 : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b1b00850a9ae484ab3b3c8b559ea79f7 : SUCCESS in 23m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ac99b1f251864378bc046c3121c21686 : SUCCESS in 32m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ede74b99170b4027855747fdb2f4b154 : SUCCESS in 15m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/393ab8939dbe4535bdf4cab599320e73 : SUCCESS in 16m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1286bca6c4243a1a3afd54be3cfe2d4 : SUCCESS in 20m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8d429edbb0643c08eca16f23cb06349 : SUCCESS in 6m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cc398fea3990415b8ce8a7d684a7d78e : SUCCESS in 8m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f9e865d2e2a148ff81ac2d001cbbe983 : SUCCESS in 7m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/db27e7f2a43d45e6baed208814d1044a : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f2cc6ae12dcc4378956156a67726670b : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69ff57ddc02b4e108b625881804459de : SUCCESS in 1h 04m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2130bf774d5d4d1494c65f4746ee6b65 : SUCCESS in 1h 13m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/542558be708c475cbbace97bfea84ebe : SUCCESS in 1h 46m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e44356ca769942759ca04469ad209705 : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":4},{"id":"08bf56551c6b3ae74ed6e947d6aa7ab8c2839004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 08:58:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e7e885765fb0b9537d52f3198c73bae74375e0b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:49:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec7b677230b47a3b2ad7bed35b6449b : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27ef0028e634d3e8a49523082c07afb : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/538c6407699b4e7fb65239ff63e28e81 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3077b39f2b994f00a95bc7237d1f1602 : SUCCESS in 7m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/42a9b85fd4ba4cf1b4d2fd71ada6d80a : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50feb4a02ec34a8e8244f6ab8ec7efd6 : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/962bec2fa5a246cdab769d017c00ba4d : SUCCESS in 10m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b8ae80eac2c4bc19417e1680d78a928 : SUCCESS in 13m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c16d731e664341e08bc897d179fd3658 : SUCCESS in 8m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/38482035f5ae4bd9b60d0950c1063826 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b035efd2e1e14713a360b4c2e33681df : SUCCESS in 14m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d648252630544abcbfbc172af9ba108b : SUCCESS in 24m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/78fc8a97b7214689a71f5129f0feacdf : SUCCESS in 30m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e96f8d7489e54dcfb4163730c86cb696 : SUCCESS in 18m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/465ac61c6e704bca92a8722a596ecd62 : SUCCESS in 15m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca9135ddeeb84d70b49290cb93b8dff5 : SUCCESS in 24m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/347598c131b44d528c8dac9cb029f241 : SUCCESS in 6m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0f3178b7dc1948d387e5238d35955754 : SUCCESS in 8m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b504f15ff5734407be45daf2a8992af3 : SUCCESS in 10m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e7fc6c77c6bb4161910aecf241f1750a : SUCCESS in 14m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/54de55aae8bb461aa38732ff5f80b857 : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d06f7728e0a437ca0d586ae1b2a8328 : SUCCESS in 1h 00m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c7d0e03f627489da35a159b96e704c4 : SUCCESS in 1h 15m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/743d28f5e4a94da1847bd2a341ded33e : SUCCESS in 1h 41m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1a1457aabd214d48a03ab189ea7dfe16 : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":4},{"id":"f9a645e5c16604533cab24210dba79baad887835","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:49:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e2ae858f344f21522b91cd1c2ef68f5194fa2154","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:51:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1972832cb3704015befbc8ce25a949dc : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b0b27ce07f22678fa14fde32b78fa6de5d6b8dfa","revisions":{"a1c5184ef45749f409676fc70b65b0185d600e42":{"kind":"REWORK","_number":1,"created":"2020-04-01 06:18:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/716491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/716491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/716491/1"}}},"commit":{"parents":[{"commit":"0cf04b1a19635f1eebf51e7ea4736af65699423f","subject":"Merge \"update constraint for oslo.privsep to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cf04b1a19635f1eebf51e7ea4736af65699423f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-01 06:18:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-01 06:18:39.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibef65d96adc52518f7ea5dc7608daddd4d1b6276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1c5184ef45749f409676fc70b65b0185d600e42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1c5184ef45749f409676fc70b65b0185d600e42"}]},"branch":"refs/heads/master"},"162c168abe4b6826fa463889f9ae359f31722fc5":{"kind":"REWORK","_number":2,"created":"2020-04-02 06:14:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/716491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/716491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/716491/2"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 06:14:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 06:14:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibef65d96adc52518f7ea5dc7608daddd4d1b6276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/162c168abe4b6826fa463889f9ae359f31722fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/162c168abe4b6826fa463889f9ae359f31722fc5"}]},"branch":"refs/heads/master"},"bf5246723c7b08be3476a8439116e951607bda8b":{"kind":"REWORK","_number":3,"created":"2020-04-03 06:22:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/716491/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/716491/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/716491/3"}}},"commit":{"parents":[{"commit":"84f03ea207731331f2c45773f08e8f21350700f0","subject":"Merge \"update constraint for oslo.vmware to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84f03ea207731331f2c45773f08e8f21350700f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-03 06:22:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-03 06:22:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibef65d96adc52518f7ea5dc7608daddd4d1b6276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf5246723c7b08be3476a8439116e951607bda8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf5246723c7b08be3476a8439116e951607bda8b"}]},"branch":"refs/heads/master"},"b0b27ce07f22678fa14fde32b78fa6de5d6b8dfa":{"kind":"REWORK","_number":4,"created":"2020-04-03 06:26:20.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/91/716491/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/716491/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/716491/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/716491/4"}}},"commit":{"parents":[{"commit":"84f03ea207731331f2c45773f08e8f21350700f0","subject":"Merge \"update constraint for oslo.vmware to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84f03ea207731331f2c45773f08e8f21350700f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-03 06:22:15.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-04-03 06:26:09.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibef65d96adc52518f7ea5dc7608daddd4d1b6276\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0b27ce07f22678fa14fde32b78fa6de5d6b8dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0b27ce07f22678fa14fde32b78fa6de5d6b8dfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
