)]}'
{"id":"openstack%2Frequirements~716677","triplet_id":"openstack%2Frequirements~master~Iad66f3be7b7cef6226ee3296acb8921d32155202","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iad66f3be7b7cef6226ee3296acb8921d32155202","subject":"update constraint for glance_store to new release 2.0.0","status":"MERGED","created":"2020-04-01 17:17:23.000000000","updated":"2020-04-01 23:50:53.000000000","submitted":"2020-04-01 23:49:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716677-1585784965976-d25266a9","meta_rev_id":"a2d6de9ddb3c2c45fc004c61c07725daae7336f6","_number":716677,"virtual_id_number":716677,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-01 23:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-01 18:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-01 22:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-01 22:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 18:20:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-01 22:14:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-01 23:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b34d9075bae8cc07f7795b2410dda070120696d0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-01 17:17:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4720d3d157c923803864873fca208c6063ae2e3f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-01 18:20:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6be1692b7f12fe38b5b7f097cbbc9ed16291738d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:03:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa59baf560614a25a6ed1ae9e4744705 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032edcbfe0cc44f390c30b8f95562542 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52f96db5181d43e8ad5aae3f65f91e1b : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a5f5f11211b4045827ef73724bfac72 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16595fce7cc243e9945ad9ec1a59c133 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbfe6fa38bb347389dc774bf7ce9977d : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb0f8425390b403fb849266b134e7ce8 : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6280a5040b6429389f89c8571d152d9 : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ba3f1e2e2e8f40fab27eede00e226ce8 : SUCCESS in 10m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fd8821550f6242d1a4556fab415a86f8 : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f3dfe99c39e34b8888f3c3572bc42628 : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/013497f1cdde459289dfc05bd435e108 : SUCCESS in 8m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f0bb229ba114e3ba3ec2a0425e7d995 : SUCCESS in 14m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/63fa884899ef4f7faafbe21cadd97523 : SUCCESS in 24m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/00c0e96f451a4260ba6f99b75fd595ff : SUCCESS in 11m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0adc2f8e5e61409abef532794610a418 : SUCCESS in 14m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f2d9da87c7554c689c76654073a01415 : SUCCESS in 17m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/db84f5f014124231aba559b2f7a2a533 : SUCCESS in 4m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/190c0c446dbe426f9387dd4c4c405e7e : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa1ce92eabac401b99d0647f7a1cac26 : SUCCESS in 8m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8ff0a8e1fae84453a484e346a3051d06 : SUCCESS in 14m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/af935a59273949419fd9429cae842680 : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25e5c08bbc36425b8f22b8f916f8ea1c : SUCCESS in 56m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d59d6ed07884b4fb1899a76b566eee7 : SUCCESS in 1h 09m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7a1f978602f422bab9db3cae4fdc32a : SUCCESS in 1h 32m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3b3fed6fd14b4d939ffccc1d8267eb80 : SUCCESS in 1h 12m 57s\n\nWarning:\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Comments left for invalid file stdin","accounts_in_message":[],"_revision_number":1},{"id":"99d145c88dbb491ef5bd8311b0c401bd5f25f30a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-01 22:14:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee9f795209041a1399f31d0f582ca4f04f6b7b90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 22:15:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b5a9b0cbe017914c31fa3992db2ad2c4d857cbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:49:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6113933ce5c4433f95f77e1817909114 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92eecd00bd2b40ec8a5e91257c3a4040 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b64303ddf8640628437daa7e1790b01 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f35f64a4843b4d0590dc72f8ac15b684 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/234c37a72fad485dbb8525e69eb46a39 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/160b31436f7b4be2896d5c405148faca : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2e3fb27fac64e4cac0c083d274c6712 : SUCCESS in 7m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/31b3d2bb052d4c5e8b036a128a03d92e : SUCCESS in 9m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14bc52269b9c46eab96796e8764cfdff : SUCCESS in 4m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/35a5db33ab554a8aaa704002dc6c1cb0 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5a5bba376844389851e75d18a08aada : SUCCESS in 7m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b7ba404625f9456eabb2cb046424708b : SUCCESS in 15m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4e8f19efcd2d441298723a4022db0dbe : SUCCESS in 25m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c2c8ebaa6e6f4802bd73ee85fc9b0766 : SUCCESS in 13m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1c1fb63e740b4a61a8a74921af2241d2 : SUCCESS in 13m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4bb57472b0b544509ca5a710387d0b2d : SUCCESS in 18m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6c2d86a5172b437cb3936fed6aa433c2 : SUCCESS in 4m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c94d1e3b2400425ba9f35c6811043e11 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc4249b8dafb4618b4101c76a6926377 : SUCCESS in 6m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dac1d2f05b5940749485e94703c2e621 : SUCCESS in 14m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cec058117ae04c0fb04f859ebfb969ef : SUCCESS in 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abdafc013db843ca94e35755ba47d5d7 : SUCCESS in 51m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/76afe5131f414cc48f498024f7bc2d75 : SUCCESS in 1h 16m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87df28aa0135476c9d4e906c9fb10748 : SUCCESS in 1h 32m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cc02f94629944318aa263c5e9a3ec20f : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":1},{"id":"291f9bf18d123b6bba66360548439f9146ae6422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:49:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec4c7590c60c917be6a1e857d3c5c44e34a33784","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:50:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47294eeab20441aabef31e54365233dc : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb561c1d804d925a2dd99dac923a3a8950378039","revisions":{"fb561c1d804d925a2dd99dac923a3a8950378039":{"kind":"REWORK","_number":1,"created":"2020-04-01 17:17:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/716677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/716677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/716677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/716677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/716677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/716677/1"}}},"commit":{"parents":[{"commit":"0cf04b1a19635f1eebf51e7ea4736af65699423f","subject":"Merge \"update constraint for oslo.privsep to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cf04b1a19635f1eebf51e7ea4736af65699423f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-01 17:17:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-01 17:17:21.000000000","tz":0},"subject":"update constraint for glance_store to new release 2.0.0","message":"update constraint for glance_store to new release 2.0.0\n\nChange-Id: Iad66f3be7b7cef6226ee3296acb8921d32155202\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Change-Id: I213c2920cbc15a3837930bef1c18d93a847712bb\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb561c1d804d925a2dd99dac923a3a8950378039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb561c1d804d925a2dd99dac923a3a8950378039"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
