)]}'
{"id":"openstack%2Frequirements~716738","triplet_id":"openstack%2Frequirements~master~I986d95e96607e0335ee72850e7d6e5f8e20832d4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I986d95e96607e0335ee72850e7d6e5f8e20832d4","subject":"update constraint for reno to new release 3.0.1","status":"MERGED","created":"2020-04-01 19:41:13.000000000","updated":"2020-04-02 03:50:51.000000000","submitted":"2020-04-02 03:48:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716738-1585799333246-e8a3316c","meta_rev_id":"55c9d8423259d273d3ad69714ab8432bf242de1e","_number":716738,"virtual_id_number":716738,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 03:48:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-01 19:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-01 19:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-01 19:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 19:41:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-01 19:43:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-02 03:48:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"929feffd4077e0a9c5a46af3cadde4df79a1914c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-01 19:41:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3dba4de7a57afb402f04ae0372a8b6a6b8a3755","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-01 19:41:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33d6b481e6d8934b7a3f3d958c3b91cceff3b57a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-01 19:43:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26a33de01529440cfc05cb191a710b548939c5f4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-01 20:42:21.000000000","message":"Abandoned\n\nShort circuiting failure due to mirror update window causing new package to not be found.","accounts_in_message":[],"_revision_number":1},{"id":"e1548970438e0b5746f3919fa4a39444f7e4da37","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-01 20:42:25.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"811225b28a4022ab1e234abca172bbe118106ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 22:20:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fad66cea46f4014b867bfd310ad20b0 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72403d52bba416683455e7196a20bb7 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9771c4e63164745b9ea74b917335d2d : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78c75a210979455ba572d408918ccf59 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb7ddc2377e34be1a51b5c0a11161fa9 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7435b00c28c4619a7eaab0bcf712422 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8fa14d62480b4a56a8a07e1225a56489 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ebbfce70bed142c58ce77df6089a653e : SUCCESS in 7m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a5b193fa38f4a3ebef1a490f57af4d0 : SUCCESS in 9m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0a786a79fbb14cb5babc40dbd1d72b64 : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a19dca4e82f0458f966b3a0f6f93fbc5 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81092417c0774df49143b6d24a1d8627 : SUCCESS in 7m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1a8a8f73902e4af5b962c29796264045 : SUCCESS in 18m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3a6f020220a349d08c02e9f2bb13c6e8 : SUCCESS in 26m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2980662e98a44cf4b0e505097454227a : SUCCESS in 20m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6818a0d584e14a9d98f62ef8860c5651 : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3888ed7c64834b36beb44b3d2be8a231 : SUCCESS in 14m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa3a0b465df4478faf91b5004d551fb2 : SUCCESS in 4m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f777347d810b4fb8b1392e12fbc3e1ae : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a2c13cefe184838a6c1859ea2d2f6d4 : SUCCESS in 6m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9b98af0c2dd041059fa7b8a3ad540215 : SUCCESS in 9m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09e187ff734f408f89abfa8abbfc7281 : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d5894949f134de6863f183c5e98b74f : SUCCESS in 50m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fa2f6c537c44eb49ec28a8dbeb3335e : SUCCESS in 1h 09m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43e5739e49a342c0b8c18107b3e2f941 : SUCCESS in 1h 32m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/09fde48e723e465490472b1a4e18bad4 : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a22f232f82666138b7ce892305219a0e07fbdc8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 22:21:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"474b855d32445117879842e05699a7bbb2527387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:49:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9511d19e61f44621a57555ae1bb74065 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c7faf1e2b04129b427c803ecd0c20f : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca48076868064c8bae782f0af747f1f1 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df28c663beb84d7293a574c50779d1f5 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a77e3c096044c718a0c8bf4df7c3d31 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13c8905b17ac453b8254fe25224e28b4 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1163e738e2f34b12b6074eeb59734f59 : SUCCESS in 7m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a8686ff892b4b168bf5f3dc59f61628 : SUCCESS in 8m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b4fe706f8a4460792cf59b05e00b29f : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/07809571d2de4a94afa1ebb6adc8ac19 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2fa75e2c8a64b11844737c04fed69f8 : SUCCESS in 6m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/30a1348c1f2443ed844fe2c70a0f2b6e : SUCCESS in 12m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d586434867494f05a9e8c03980981bb9 : SUCCESS in 29m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/57cb1082a3dc413596d22c5fd51f0132 : SUCCESS in 11m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f8b4d7ceb0b74d7bb094dae80c385e68 : SUCCESS in 14m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6fdf7704e4a48698e14bf4b01a38c81 : SUCCESS in 18m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ca3276777c8d423291459cc4cb3ad39d : SUCCESS in 4m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8ddda22cceda4e9c931767252c739fa0 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9428bb983ff47ac8bca9ac41b2fc596 : SUCCESS in 6m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cff88b4b0351429fa770cdf1f7ae17f3 : SUCCESS in 10m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/23b46e0b2767442882ee5c3fdcb37c06 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8503a115eb8f4762b26bcce3ca6a8327 : SUCCESS in 52m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec515f26d0c845ed86eb2df9a5ad0812 : FAILURE in 58m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f15a1e0308e64bf7a5437f0f914ad5d1 : SUCCESS in 1h 17m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1b1f7bad5d084efd9ab22a424eec62cb : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7828fbfa7b0c9f9fc947b00ade5ede7e6e07f425","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-01 23:54:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95e007f039157965178034e20667de1775a301a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 01:23:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e9fb0ea7ea94caf80fba005c2abdadb : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eda80a2ec744bc981928d90a9282766 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28aa79d621dc4a0b99c65217b423e914 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f945d0751fe34c30850d3b8e912028e1 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f6fa51e4a8d4c72bed592d11c99f726 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87fcf0b487564c1c9321b5aea4eb88bc : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d241f0451a8f4fcb8ed2504efac9771d : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dcf484d595eb46bea6d5911ef3315345 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8dcebd92d1c4670bf421cab2a1b08e8 : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/244a94dca6114b6fa1d2fdf3aae790dd : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4cffb2498a07438ab254e60982820dc8 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/28eaa71950054d2c8dccfd36cf63d72c : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/586255bfc85e4309849214d0ff3dd9bb : SUCCESS in 19m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/72939f0905d84de796537c5f3b2ee228 : SUCCESS in 30m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d0d10cdbb86446e68bf057c695005f89 : SUCCESS in 22m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fbeecdbcfa5743d3b6ba71c61af7ccfa : SUCCESS in 13m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2774f692a6f9499d9a9387534ef8acb4 : SUCCESS in 19m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/455e2a368ff34ae8a61fb4143ffa1383 : SUCCESS in 4m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2f25085767d54d5bb07771b559f72218 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/adad521eaa564eeab3dd03bb377f9d69 : SUCCESS in 7m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a1738ed5aa2c4f6382a5b253b93095d6 : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c55022c4aa65473c824be3341c439012 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17e59622d4584e5e90d0cb1071793b56 : SUCCESS in 57m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c24f774079464241b07467cb1ba88ad8 : SUCCESS in 1h 10m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b05542145c447f4afff7f3b5b5940b0 : SUCCESS in 1h 27m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4b49dd2edc1d463ab5fc4ee8bde4645c : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c38927ae748065569b7f10f81100988b2692c2a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 01:23:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ca8619133a3b7f58571c2f62858e26aa0e0e592","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 03:48:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d4dfc4548055f9b6fd7a969ed059783f7ccf849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 03:48:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28096d4c615c4eb7ac38ac730648cf63 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b32504d1864136b4382ce1f1315090 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d231ccde2c94a578368847c544a7355 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1408c0d892844c498e1899b21aa8918a : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70e51059f3d24ff28c57955399f1bb61 : SUCCESS in 6m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82a21f75d0be4a6288f7515a8204d1fc : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7681ba736db84b938b3bf3d14502512e : SUCCESS in 7m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d94a736c032d4d0fbcfd921ed3e436e1 : SUCCESS in 10m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7f7b18cd7b23467289352a9bc1855f48 : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54af6eba439a482c97a3ab7619292ee8 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c163736b758440298791efaa2ff6f236 : SUCCESS in 12m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ab9fbca6e1c442f0b68409b6d68edce8 : SUCCESS in 21m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/330bf35f0cdf42ddaa4ee5bdec45edf9 : SUCCESS in 28m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0317217792a14a599bc0ddb04547f595 : SUCCESS in 17m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/93aa4a70db3a4b07a6d1ffd17631d548 : SUCCESS in 15m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30a5079de06343df92cf1b0a273bbd22 : SUCCESS in 16m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/426d919d6edf4874b80c50e26ff3fe62 : SUCCESS in 5m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9160e71d5adb4dde96f84d6e9b61fecc : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c5cc24fb53d4bcba73a11e302776c55 : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/68c2c77b512c4a91b2449fbf8d5b2626 : SUCCESS in 15m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0b03652231f34b559b5c874490e36925 : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73e530b17fa84db2a17723861af5f6d5 : SUCCESS in 48m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc71bef914434d0bb68ad6fe3841f203 : SUCCESS in 1h 10m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d175e8f8d50b45c7b150dcfe49661366 : SUCCESS in 1h 13m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/852005efdcb5489aaa43e2420f5c0d14 : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f66b77c23634c535572582bda7d322ae0c9a355e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 03:50:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae6b5d1844f42d881c512281df3651d : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4101c6889638f65c05ad8f941290dd361f98cd3","revisions":{"c4101c6889638f65c05ad8f941290dd361f98cd3":{"kind":"REWORK","_number":1,"created":"2020-04-01 19:41:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/716738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/716738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/716738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/716738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/716738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/716738/1"}}},"commit":{"parents":[{"commit":"0cf04b1a19635f1eebf51e7ea4736af65699423f","subject":"Merge \"update constraint for oslo.privsep to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cf04b1a19635f1eebf51e7ea4736af65699423f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-01 19:41:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-01 19:41:11.000000000","tz":0},"subject":"update constraint for reno to new release 3.0.1","message":"update constraint for reno to new release 3.0.1\n\nChange-Id: I986d95e96607e0335ee72850e7d6e5f8e20832d4\nmeta:version: 3.0.1\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iaa88eb6a7b673bbe997facaa890a5b52af3bc1a0\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4101c6889638f65c05ad8f941290dd361f98cd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4101c6889638f65c05ad8f941290dd361f98cd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
