)]}'
{"id":"openstack%2Frequirements~716945","triplet_id":"openstack%2Frequirements~master~Ie0e236e792938fe29873068d07679c6b1f7fcd68","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie0e236e792938fe29873068d07679c6b1f7fcd68","subject":"update constraint for sushy to new release 3.2.0","status":"MERGED","created":"2020-04-02 11:09:51.000000000","updated":"2020-04-02 18:39:33.000000000","submitted":"2020-04-02 18:39:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716945-1585852773728-cde04409","meta_rev_id":"497b4e2b1def6cc2ea3a371f99ee3fafa86290ec","_number":716945,"virtual_id_number":716945,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 18:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 14:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 15:39:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 15:39:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 14:38:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 15:39:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-02 18:39:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73dc32acb17f877f231c853173b0f90d0a65d1f4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 11:09:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6f59928ec3d131bccf240d4fd8ba25823218ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:39:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0690eb4e7cd460aaee011e6d960749c : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d76355990d468a805630af41bed989 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1966f42aa3db42dab6e641e6280ea105 : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13a990365bde40e096c6922ae2927566 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/994133abe03c4232bd911c89a24376ed : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b06b40a84164d4e92e766367ea756de : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b1ac913921149f7b1e8735455d88f4b : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b74e03f3c784404aec09bf03a0ac9a7 : SUCCESS in 9m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/246001333fcf4ea2b451cb10ef12a2d0 : SUCCESS in 11m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dc865f484796442a84a00826da958e1b : SUCCESS in 7m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0bccbf9ebc424797b67b40d99b046d58 : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1ab430825bd47dbb2d50224a00abc0c : SUCCESS in 8m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8ff28759d61243c4ba589fd4e65b5e7a : SUCCESS in 22m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d679c4e42df4696a02c551cc9b21304 : SUCCESS in 26m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d9a14ca157324f28804bc2be9e19327e : SUCCESS in 18m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d01e5eb435034adea152f04a9b8d9116 : SUCCESS in 17m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d2be7124c594037abea382050953279 : SUCCESS in 19m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d169c4830b934bc0933d7d3d9b885b56 : SUCCESS in 5m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b2902d569d6645e6947e53add2e83770 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c77fd02999f459d9207cbc7aad5d2ce : SUCCESS in 8m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1aa88ab32a184441b5bd3a8b908ebe5c : SUCCESS in 12m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fb5fe5fa9bf84ad391af46881c926649 : SUCCESS in 11m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e35fc54dba90408a9eee1b02e79855c8 : SUCCESS in 50m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cf72c239f4c43f6905736294c460041 : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94f06c7fcc4640998bfba4232d9ac93e : SUCCESS in 1h 18m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/080fa9c6ec594e4cb46553b992d782f6 : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3bc73740c0b46cab18eea8e11331d735f2a5b58b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 14:38:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1165b94de4c7526cb24a667e653713f9a1a1d188","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 15:39:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4915feccacf62f053cab6d2ab3382f0a592cd28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:20:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a399ca9db6a8e08819deb5de621d37e984403188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:39:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8366512e7e6441818e69fce4e8946c35 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41f381688b99493cb5a6c7497d974bd0 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9811c72b68084354a8f78d5821f97082 : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba8f145795f147a7907ba9d7ed811526 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/611162f078524d3fbe87a39038a75db5 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c69bb5dca8f4c8fa3958078f029795b : SUCCESS in 7m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12e5b079a7ee466c892ae1171d222034 : SUCCESS in 11m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa8d8c0f0946447cb7f554d8460545ea : SUCCESS in 11m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fa976cb0e5b249bfb03fc089d1eab270 : SUCCESS in 7m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b19e994fd906477891a5183e354bcddf : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c0dd6cd8124458181613f13180984e2 : SUCCESS in 11m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95a49c7e19ec4456b75c9ee93d15e472 : SUCCESS in 19m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/283cbfc73f0743d4ae863f0184e338dd : SUCCESS in 29m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4d0e92035f6342fea768018af70aeaa7 : SUCCESS in 13m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/08b0311a8eba4301bf6f8bb29f7a2c53 : SUCCESS in 21m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc8daaa417014bcba62705977f80a6b7 : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bcba163008ed4ffcb0fc269919c74256 : SUCCESS in 6m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/86fb3460b4bf4bf8ac65367cbe5156c5 : SUCCESS in 7m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9156fec7af2b4b30a246179708a85d25 : SUCCESS in 8m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ab2634ed0a7445f5be98c7130c72e9d1 : SUCCESS in 12m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/35f8765f11f14f1d9108c3b5253d77ee : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/058c1e87b3c449cf8eb75aba669fd2e4 : SUCCESS in 1h 03m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee59fe97b96d409aaa40fe59d145202e : SUCCESS in 1h 18m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c68aaa8d70254f35ae521bcc217df0b4 : SUCCESS in 1h 18m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6c3f7afb041f4524ac1dfe5525e722b6 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bca22cf7afa373b1adff3fa16b73524bd2f86a64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:39:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71dd6253fcf9bb77016d6ad39bb7c7b80bceb260","revisions":{"71dd6253fcf9bb77016d6ad39bb7c7b80bceb260":{"kind":"REWORK","_number":1,"created":"2020-04-02 11:09:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/716945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/716945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/716945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/716945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/716945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/716945/1"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:09:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:09:49.000000000","tz":0},"subject":"update constraint for sushy to new release 3.2.0","message":"update constraint for sushy to new release 3.2.0\n\nChange-Id: Ie0e236e792938fe29873068d07679c6b1f7fcd68\nmeta:version: 3.2.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Commit: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Change-Id: I8aa5ee8cf9d6c5f51d917cdcdff9503c06eab043\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71dd6253fcf9bb77016d6ad39bb7c7b80bceb260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71dd6253fcf9bb77016d6ad39bb7c7b80bceb260"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
