)]}'
{"id":"openstack%2Frequirements~716946","triplet_id":"openstack%2Frequirements~master~I31a572ceb367a50693cff839e1eaa4f6ad7d9b21","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I31a572ceb367a50693cff839e1eaa4f6ad7d9b21","subject":"update constraint for heat-translator to new release 2.0.0","status":"MERGED","created":"2020-04-02 11:10:58.000000000","updated":"2020-04-02 18:43:28.000000000","submitted":"2020-04-02 18:43:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716946-1585853008273-9820b1aa","meta_rev_id":"1df932b845ca6f444b965a4b9ffab0c8b82ed6bb","_number":716946,"virtual_id_number":716946,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 18:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 14:38:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 15:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 15:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 14:38:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 15:39:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-02 18:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e2b9e55a2564a6be2a710956f6df23159a67d1d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 11:10:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4771241165a302eb9a14902985ca4d7351e7a435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:56:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b405affe2cd24cf782bdc2d6ede8aeec : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35f64686739144c8b1507f5e4b4359dd : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1577341bc5ee4dac89ef7d5a8e9f82d2 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff8f0f6a98484838a7240eaab0ffb4f8 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/705bf86e3a6b47f0a64b0c7f42af779c : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45f99e037701431a988b9e7c87c1fa3b : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61bbe4d8af9d4a2e9cd31e81ef853646 : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/290f9dfe78ab49e4970839e0d0812548 : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a0d28bf119e1462e9c6b0936210a46dd : SUCCESS in 12m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/faf5ad97f331401396e862f1f892697e : SUCCESS in 7m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/16a0aa86200b4b38923186784e1e64d7 : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/798ac483d6184e02aa691cfb87e5d62d : SUCCESS in 8m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b785f77e9ff2487c894f90c4480de8c7 : SUCCESS in 14m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/74279a7c45944655bf01dc97c6bb4c2d : SUCCESS in 26m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/05aac8a8ea174bb29edc77a90dfe8556 : SUCCESS in 17m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/492bbdb350e64c309e05c813e6b2cbaf : SUCCESS in 17m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ed59378e293477ea5daaed0fe67b395 : SUCCESS in 20m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3b1868b8fe0a44f79c2d0460a4dae6da : SUCCESS in 6m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/42527d5c883a44aead618f2c1adfa8a4 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12f3a5ac1ffe4d8f8fbbb1217ba87fc6 : SUCCESS in 8m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6cb63acbbc4542b5a444575d2b5b0edd : SUCCESS in 15m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/efbc14ce57b34862bcb52b61f1f9a13c : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7905d61b051450896101f9f1139fba6 : SUCCESS in 55m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ca678b96ddb4a4f91f413b1330859ef : SUCCESS in 1h 03m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fef9e7fcfeac4783bc61b5789e6b5f3e : SUCCESS in 1h 21m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1ad496f046994d54ad40c421548bad3d : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0ab5ff7a92021eb650471bfe83b0bf72e34999e1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 14:38:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bbace70fae49e20ffc2d1d0642032cf583edb207","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 15:39:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"117a3afb5dd6432bffe6d4e6384c31e5fd3b5d97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:20:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fecc7d7024d6e9e7f45ceabb593ee7fa27448353","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:43:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174f0b3b6f7e49f39a58a3b90ea0c2f6 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aff24d1bcbea44788fe85a6078342890 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6359411bc00d4bc39d24e1a2c0e90bdc : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/418f0e88ff3344d4bf8c61a998c139ba : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85b74b04deb549a297016cb4169e4a23 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12af6b18938e4077b2627402ec2b40b8 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/43d23ca2a39644cdb0f9fe1bf52cf449 : SUCCESS in 8m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/182ee2ddb33149bdb13c1563b2b1f23c : SUCCESS in 11m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/71d8330dc20e440497e889cf5c10cf12 : SUCCESS in 6m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e4e261c7b13344739a3053f8b4aeb0f8 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c38a1313f33046e6bf5fc1836829fe4c : SUCCESS in 10m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28f55c9e95ef4fb8b9b9bda5497854b5 : SUCCESS in 17m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4275f5996cbd4af7a2eab3826403dbf9 : SUCCESS in 24m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/703b7ad750a64660a79933bcfd0d95b5 : SUCCESS in 21m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fd8e4cb60480462cb5db9230d45994e6 : SUCCESS in 19m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1703ecbe6fcf4ae58abd63a335edf011 : SUCCESS in 17m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f2b6cbad129749a18ae7dc2523b8802c : SUCCESS in 6m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b596348e65fc4b17a0db0aeb7897230c : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c4cd8a397784d208557ae711ac9fd66 : SUCCESS in 8m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c5124dfa37864554a9ab675d5713e226 : SUCCESS in 10m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee53d83bdc944c838c76387df5f20dce : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4e1b1ef23964fad98ececa0dc2626fe : SUCCESS in 58m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/416c8614f5bd4813b849cf6f933518a4 : SUCCESS in 1h 20m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4709c2fb50ba436484215c1bdd1528d9 : SUCCESS in 1h 20m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/38f37ce41ef84143bed734e360c36a63 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":1},{"id":"89c565a2d858c74491a6ca3976a4fd5aa531bb4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:43:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c13f48f8eb592853331cd5e619df9b47fa79c08f","revisions":{"c13f48f8eb592853331cd5e619df9b47fa79c08f":{"kind":"REWORK","_number":1,"created":"2020-04-02 11:10:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/716946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/716946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/716946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/716946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/716946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/716946/1"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:10:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:10:56.000000000","tz":0},"subject":"update constraint for heat-translator to new release 2.0.0","message":"update constraint for heat-translator to new release 2.0.0\n\nChange-Id: I31a572ceb367a50693cff839e1eaa4f6ad7d9b21\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I6d79952aabef6bb5d21fd932eeca636fd83b48a1\nmeta:release:Code-Review+1: Rico Lin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c13f48f8eb592853331cd5e619df9b47fa79c08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c13f48f8eb592853331cd5e619df9b47fa79c08f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
