)]}'
{"id":"openstack%2Frequirements~716947","triplet_id":"openstack%2Frequirements~master~I915be3ea95d294e5278a6c472f1d5ed442b9ad13","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I915be3ea95d294e5278a6c472f1d5ed442b9ad13","subject":"update constraint for metalsmith to new release 1.0.0","status":"MERGED","created":"2020-04-02 11:11:26.000000000","updated":"2020-04-02 18:45:39.000000000","submitted":"2020-04-02 18:39:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716947-1585852775691-65716cc9","meta_rev_id":"6190e27cd4e6ef2c81515d68ec2b390c69bc0880","_number":716947,"virtual_id_number":716947,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 18:39:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 14:38:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 15:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 15:39:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 14:38:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 15:39:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-02 18:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79cb1cb8fc8a05e231ac0f5a12db5ecc23783aa1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 11:11:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9adeede7c38d06acb8673c4c8b82adf923f0863c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:56:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de725925c86e4ab49b1ed48e715c074d : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c38af7e9a1ea4ce6bf21bb9403a8a856 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64756271a0ee4a5cb6ade161e1c715f2 : SUCCESS in 7m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/628850ed6e1145d9bf1c6189cdb3add5 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51fd095c30104f9d8fc22d3cc386cc60 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ec284ad265e4dc3b99e663a7f7e09eb : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a1b6b51803a419b8df3669a0eb7fad8 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9eec962b6994d979399529c6a4a9976 : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85a0a15c02a2436a9fc10e6f095fa509 : SUCCESS in 11m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/60893a033ec94f34b5963d38828d483b : SUCCESS in 6m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3131444a73f242ff981098c9f70e258f : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/479c07ed88274dc0b95a17d28e468a93 : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b4ddc86bf12444f19fbf3b36781f19b4 : SUCCESS in 15m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8032742f992e4c81a3cc595085d8beb9 : SUCCESS in 26m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/274e5ce1b4444b8180202d5ceffcaf0c : SUCCESS in 13m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/73a67c498f88493dbf4bfb3b400cb8d2 : SUCCESS in 15m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5313305858b94552917a221050720c60 : SUCCESS in 23m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9dc1c7f444214763b94a864a2c69dc70 : SUCCESS in 5m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5068e2b2c96d40018edd66b3858fb604 : SUCCESS in 7m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2684d06c8edd47a286be4f9cf25c0f63 : SUCCESS in 8m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b18e7e2c6e964ae9b8bc96a672f1cfce : SUCCESS in 13m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8f2b8781d6af4edeb7f631262edd8e97 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/701b62097db64c37a57672a1a4a592b1 : SUCCESS in 51m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/711bb5592ac94017ae221bb50dd8f29a : SUCCESS in 1h 06m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a6da9fd35bc4e69bf1478e2350b3785 : SUCCESS in 1h 22m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a534095716e4b33b3956f7d782b0293 : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":1},{"id":"442d886aaa8b26f4074aa2e8e5405f998febfebb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 14:38:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c01e3230a3877264c01f6435ec551f12a6facd9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 15:39:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a622609e47a77e7d6ddaecc1c7e92dd771b53a2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:20:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b8d4801fbd569994bc02b6f5d94e40e626ba66c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:39:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c2ff343b706e3d4195da8661f4eea27e4b595aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:39:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133de1e1b4f6410b86563d26cfdbeb25 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca653d9cbeb340a28d2e2d75da7831ab : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbf955cceb09498bab09b7d08f3c1ec9 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c33a967e1df04151a5a8dcd37b55a510 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebf331c96f544a5eb724ba20a2df90e8 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3faaee1051b3443b8438d25aa4ef31a3 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b2e1ee859a24c72993d504496922d22 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cbd4ac9466c043c6adceb2acc5668e4c : SUCCESS in 10m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/68edc50716ba48eda37a74c8f6d4e83e : SUCCESS in 6m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/20fa1b888aba4eefbe42fc8c89245f1f : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e7186dd29e747239ad77436b0d37d9c : SUCCESS in 11m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/44a2e5befaa848fcb24a2babed34450a : SUCCESS in 15m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a96cb1d889fc4fd2b623ae30611ac220 : SUCCESS in 32m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c8fd05a7d7524643bf42368ee7fb1c29 : SUCCESS in 14m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8bac3ab6dcfd42f290fc758b92703c1c : SUCCESS in 14m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd5607990fbf45378fcec095758f992a : SUCCESS in 21m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/56622c7cd351430db15f26d1f3f31fc0 : SUCCESS in 5m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c6e05856f9d04cdcbb235775814a06b5 : SUCCESS in 10m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f6f8b6bf94e457bb968155ac352bcdb : SUCCESS in 11m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4f579284960f4de1941014a9122c5596 : SUCCESS in 10m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/959d6d699e364120bb4f2d6478b07bee : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/952aa92b2f25493594c3675b85815738 : SUCCESS in 50m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d05f966d3ad64ee48b5c662c44c9854e : SUCCESS in 1h 12m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02716ae18b0c4620810486e95d172dc9 : SUCCESS in 1h 18m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/46b7769b958f42e394766e79dc5bf01a : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5937ea976039eeee546303cba64b51b5d1c3c689","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:45:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3356d734cee84f719bd54456cb5543ba : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c278266b50bfdfe2e44dc8bddd27beeebe24e5a2","revisions":{"c278266b50bfdfe2e44dc8bddd27beeebe24e5a2":{"kind":"REWORK","_number":1,"created":"2020-04-02 11:11:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/716947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/716947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/716947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/716947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/716947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/716947/1"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:11:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:11:22.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.0.0","message":"update constraint for metalsmith to new release 1.0.0\n\nChange-Id: I915be3ea95d294e5278a6c472f1d5ed442b9ad13\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ib4852b4e67170020e5acdf04f23712ef02dbb1cd\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c278266b50bfdfe2e44dc8bddd27beeebe24e5a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c278266b50bfdfe2e44dc8bddd27beeebe24e5a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
