)]}'
{"id":"openstack%2Frequirements~716948","triplet_id":"openstack%2Frequirements~master~I14682a73471a2c5c719ada10d2830f68ff59ba87","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I14682a73471a2c5c719ada10d2830f68ff59ba87","subject":"update constraint for octavia-lib to new release 2.0.0","status":"MERGED","created":"2020-04-02 11:12:06.000000000","updated":"2020-04-03 00:30:42.000000000","submitted":"2020-04-03 00:27:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716948-1585873679194-afcf8495","meta_rev_id":"5969db3c55da8d978fff59344436df593f5a06f3","_number":716948,"virtual_id_number":716948,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-04-02 19:42:24.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 00:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 14:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 15:39:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 15:39:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 14:38:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 15:39:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-02 19:42:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-04-03 00:27:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de7b770cddcb2e8525228833016971efea89b415","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 11:12:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b06232aaefeb036c54a1a08eb3a18acc90842c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:51:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9abd6943978d42dfa65b3fa2197535f3 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3542e0bdc74849a6a3b4dd084e1aaa9d : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb34c876619f4e3fa579d5b9d476e479 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f50ef4190464da4bba1e56fb0e30cab : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7c52b3ee8440f0acb5181c3921b405 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90d6ee6bdf624e94ac31d9f38cd68780 : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b25ca616a7df4f3d98dc7d0522c273a0 : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/477d2f42646048798b98ad3cf0aa7c41 : SUCCESS in 9m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/17aa0c3074414ccc981f3a22dd416631 : SUCCESS in 11m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a05dafc5d0ba4915a5cf1a1034b3f76b : SUCCESS in 4m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/abd3ff5eefd145e3bcd5fd1796c422fa : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eed6d6f6edfb455ca920ebf46e06f066 : SUCCESS in 9m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5c0249e8b1474bfb84a290f613cdcd90 : SUCCESS in 26m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f4ca36aebe74437ab721c7fb1351306e : SUCCESS in 26m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c70b7849499d4198be70441e4014a3dc : SUCCESS in 21m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3b85ec8a1e754014b717bb876e1f2234 : SUCCESS in 23m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e55745d2d0c24350b368fe66e291f0ff : SUCCESS in 20m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cb52086ff2ba424d8bc4830537e7b43a : SUCCESS in 5m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/89b7ea59e7044c41bfbc428ed809741b : SUCCESS in 7m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9440ca393d845aa9aff31cf7325d00c : SUCCESS in 9m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bcb05d96b59d4d7380d8a350f0f68f92 : SUCCESS in 11m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7ddafca448364075be228674cef54ec4 : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d53cd98ef12340308944f16f3dddc87d : SUCCESS in 1h 03m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9594a8bfbea746bb80a2feec0d740243 : SUCCESS in 1h 09m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbbd3b7fa6b0471a963bea904456876b : SUCCESS in 1h 18m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8bea61aca20449e38201e2ca8ed515ea : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2b9e354c8dcaa5cf5d2e1eee2b900fb23075f98a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 14:38:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f2b059d1a3bcec89a9b206ed40af379578ffffd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 15:39:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"869f68b8ca0f53d175652cbcabe0cc310b7da31e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:20:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35d588c6a62319887a3b9c51708e2d8cda9f88f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:43:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04ec0c848c62418f9a160a5d24abbff1 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ceffed6eec4857a923e3a8bb37ba49 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5736d2570fb346e6b92bc140945cc67b : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6359da58b4e4167b6396ee6aa233ee9 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acd22498f4fe4a9fa925e3b762930861 : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/875c329d49ef460cbb6c54d8f934a1bb : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c61971f837344be9807c9a7d59903076 : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/06afc4610ac04ba485a02da34e2cdcc2 : SUCCESS in 11m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2cd99ecd92f54146a6a352966f288bd1 : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/08859c55ea034f419dd34250d49e099e : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71f628a75fe84cdc910c92dcfdb29882 : SUCCESS in 7m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c3dce7963a6428e881c7d5f3e0b0b5a : SUCCESS in 13m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e1701712143540f99b5e360d1e4a0470 : SUCCESS in 28m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6c41a3d82086405ba7869a1f9da8a74a : SUCCESS in 14m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/711a4f0ecf9e4d129a7a4901a573c944 : SUCCESS in 16m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a5341cfce704a3d975b5e9919191ba3 : FAILURE in 21m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/52ef6616efca47529c617e462bfb0e32 : SUCCESS in 8m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/35b738cc068749a6bd53c6ef89e9916b : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d939a1aabfc94541a7ed46d4488b6001 : SUCCESS in 8m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e1fefe439cfb443e832138dc706e0b4f : SUCCESS in 11m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9f09ba43f47e43428574f4bb28975e3e : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47f8b28705eb45b08f89b2693893db4a : SUCCESS in 59m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eedf66fe6215474fbb54551c8da65e4a : SUCCESS in 1h 07m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/199fa145187145e1bc1361e540651e9f : SUCCESS in 1h 26m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/db8b33ba684247488f0dec8b70f5fae9 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f4d3d31f5c0d047d527ce1a087d32e16ce44172a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-02 19:42:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba8dd13a49a144f4e745f9cfafc13764c70fffdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:42:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e968b45172741c7a1b22b077a7d92dc : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3173040b1e4b92bfff54f9f8cd8a79 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f379e09ad580494b8238d5857f75333c : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ef00b4384264e3d8fe40424d1c30da8 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/559c2555a27e4462a82a2a46419578d2 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1014144449b743888598fda8a6ed51ea : SUCCESS in 6m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd37a475c58e4e16aec25b8fbec2b495 : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f51b70ce7ed94d3abf364fd57dd4d1ad : SUCCESS in 8m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3be04bdfe199488b8f844ef973dc8f32 : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5eed6cdecd24ce583923f834f9af7c7 : SUCCESS in 5m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92e405dd8e4949908d7eaad2e1d0a946 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae8665419fb14909a0ecc55f4be3310c : SUCCESS in 10m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3d9f50e259c944429eb0de567a499415 : SUCCESS in 21m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9351a0c89e574d04a3e76527cc02e248 : SUCCESS in 24m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/97abf46de0a941899ccdf5b607641b71 : SUCCESS in 14m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a42ee0855550498f8df64faed050e694 : SUCCESS in 19m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12c6e55645c74c59962c54fc98952b37 : SUCCESS in 19m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e38d143d70e045898985017047a3a828 : SUCCESS in 6m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7f86fe79f5ec4006ae5f43f901751a96 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f81d4d2c95234fddb907828e8f79b2c7 : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c59c903808f940218ee3329b7b14a070 : SUCCESS in 15m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/990ea3464969421ab63b8db5669b6b24 : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73a994dfd14544e9b120d9cb2f562f5c : SUCCESS in 56m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0dc61efc9d8c44dbb883bab8d77da210 : SUCCESS in 1h 09m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f6f91d546c149c488b767e77f26a63d : SUCCESS in 1h 22m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cd07a7231b7a4cb89eea677c626c201a : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":1},{"id":"42343bca8567593f3659a5e44726cafcb243529f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:51:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64d2297322aa96664f069314237c07574bcca3e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:27:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2093314611541b7af884c1516e14764 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5ff175c2b049a1846102c62222b5bf : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faa5e7de7b394d4fb8fbbaf334a131d3 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b39582ccd3324d0286ab2be820df7ed9 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8d77a3393974de9bdd7d278e545b1ba : SUCCESS in 6m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50265db8c7ce403482cbc434ed5f82a7 : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b7107b31b2a64a6dacd5d690e2f1f945 : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f353d0863adf4ed994d1b7e9ab968d71 : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6b88845a698b4b4c82063dfd14fc6f1e : SUCCESS in 6m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5dd7cc25344841ddb5c63450b5ea4360 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b19826748ecc4dc3ae8dd85136147f09 : SUCCESS in 10m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e6a4a7e405af4360aebfbc7b3af65339 : SUCCESS in 16m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/76c1068bf58b46489b6db8eefea7f5d2 : SUCCESS in 27m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c537516fcc424e64a3f0a49ce2bfcaac : SUCCESS in 16m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b8f54f8e9553448d983d8da238273ba5 : SUCCESS in 15m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/caa9dd576ace42ecb2a31afde56d7e73 : SUCCESS in 20m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f1c6c20e623148ce9108b328a450bef3 : SUCCESS in 7m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/217a13815fb44473ac8b1cced388ad35 : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad59394230734147a340fdb9c7508241 : SUCCESS in 9m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/28d59cfbee4d47249894f11a7dedfa0e : SUCCESS in 11m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/895f69da114a4c1ab9a8306c9d3f6558 : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36e7dd6741ae433c81ef6e9b2ba5749f : SUCCESS in 53m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8e47828d15f436e98441d75b5a79c2d : SUCCESS in 1h 08m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c770cd62c8ae4471b35f92740c326759 : SUCCESS in 1h 26m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6480ec24221c41c3a68f686bced1acad : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5559704265d5c5e1fd1298c5b6f10ac8eb7007f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:27:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e4e07a1871cc587e4f6d35151ffb0ad3b26ad8ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:30:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b2d489cce7413fb5f4f4e1026de51e : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"938eadf15f9ceff89befec497034c50fd47a6b0b","revisions":{"938eadf15f9ceff89befec497034c50fd47a6b0b":{"kind":"REWORK","_number":1,"created":"2020-04-02 11:12:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/716948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/716948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/716948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/716948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/716948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/716948/1"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:12:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:12:01.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 2.0.0","message":"update constraint for octavia-lib to new release 2.0.0\n\nChange-Id: I14682a73471a2c5c719ada10d2830f68ff59ba87\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Change-Id: If9b80cb8bb4893fd9d640b229df3fb54132aac0b\nmeta:release:Code-Review+1: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nmeta:release:Code-Review+1: Adam Harwell \u003cflux.adam@gmail.com\u003e\nmeta:release:Code-Review+1: Ann Taraday \u003cakamyshnikova@mirantis.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/938eadf15f9ceff89befec497034c50fd47a6b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/938eadf15f9ceff89befec497034c50fd47a6b0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
