)]}'
{"id":"openstack%2Frequirements~716957","triplet_id":"openstack%2Frequirements~stable%2Frocky~Id632090361eb93dfa2f5f4fa735e671d8742ff9c","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Id632090361eb93dfa2f5f4fa735e671d8742ff9c","subject":"update constraint for tosca-parser to new release 1.0.1","status":"MERGED","created":"2020-04-02 11:28:22.000000000","updated":"2020-04-02 19:34:01.000000000","submitted":"2020-04-02 19:34:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716957-1585856041351-f04f693c","meta_rev_id":"35c07f48f5d6b7787816ff6bfcb5e7d8badec084","_number":716957,"virtual_id_number":716957,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 19:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 14:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 15:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 15:39:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 14:38:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 15:39:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-02 19:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef79db36afdbab68947f1b4473b1dc71060b82ca","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 11:28:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ea1e68f514c99b1f66dce9c618e4f841fd40521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:25:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/49761fc9c8a14ee8ba3da28a44e3bf1a : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14653f6f6e224a1db72f028220ddd788 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a2c5acecc734ef5ad230d4c8be4e532 : SUCCESS in 6m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/41a2362abc3040e8aed9a006d6e71ad4 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffb2b0438dfb498492157dc0f275ae42 : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0d4d2e231e094ecfbf8c49ffd743dfcc : SUCCESS in 5m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0e4955e9cc2f4390b7a59fae58475ec1 : SUCCESS in 8m 57s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d69b0c41af7c4e95b04adb9076c2c58a : SUCCESS in 8m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce38496eece54a20941e511a358ad506 : FAILURE in 8m 34s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1cb9593d738d4c07a28c0ab05c1ed681 : SUCCESS in 13m 03s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/700500f4bf6749d79cccf709cde51b6a : SUCCESS in 11m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4f88261d70c644288bda7465cf9e011e : SUCCESS in 6m 48s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/06afeb3d21d54159b94be7a0ab504d75 : SUCCESS in 7m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3b8d2ca7ae064c7e848229dd33a7dc93 : SUCCESS in 12m 07s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/77b7c7e2adfa42519e97fdd2492909e4 : SUCCESS in 12m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/735a6b35245f45fbade919071fd4245f : SUCCESS in 19m 48s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/84f2c6248d2948ad802859c83e245529 : SUCCESS in 18m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4647ee5cec5c45a2a491ca78b4ff691a : SUCCESS in 21m 57s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4b791972e94743e6a998caf8baed7f16 : SUCCESS in 23m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c74d58013a494faea72e98d138ce08f1 : SUCCESS in 13m 59s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/23a03790a7524a368cafccd03af279b0 : SUCCESS in 16m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04478d82e6ce4dfaa5fc9e0a9dd201a6 : SUCCESS in 20m 15s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/ae9a2befae50466d889087a13b46da4a : SUCCESS in 7m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9e6959aa247a45a3a1c14bc324299530 : SUCCESS in 10m 05s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4987dcef745e438c8076cf04fe0a349d : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fa416c5fb1842a8b21f3dcc26ad04ab : SUCCESS in 5m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/160170381c8b4415a4ade1c86daba042 : SUCCESS in 1h 06m 52s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/2706b7edcbcd4619ba916277ed14e8d5 : SUCCESS in 1h 32m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6b9f9a27dd7a4376bd2512a5e25a343f : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3a180ef409167326bb369406594998be7fe3b752","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 14:38:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1af9e825cb071a9b3e571ddc3c4c6be0d121cb6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 15:39:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bae66893838709c9762620dfdb7d792308ae3c2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:20:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c22fe0a1a8fc24147185e7694d5171d23da72b89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 19:34:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"33f7f16413af06f341885fe28eb125dcbd2b408a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 19:34:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d450aa52d36e4ceabc02e043c48902fc : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/402b3fb8a3c346cabcfc42b8c68cbe59 : SUCCESS in 5m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/03013c1fcdf54db3a110cfbf4e64b4cc : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24e800c0a71240fea8374374a79b0a37 : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7dad25fbb974fe2b30de3ae31492968 : SUCCESS in 5m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fe4ae1503ccb426f8c214260ceb574d3 : SUCCESS in 9m 40s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1a87140f15054036a94755312e011c17 : SUCCESS in 10m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b7c60534c94c4dff9c4cda1777b81af6 : SUCCESS in 10m 45s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7a6bfb5ad52643e1975271e021b77a8e : SUCCESS in 11m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0ddc33b13db84b2786a7bebda600cf74 : SUCCESS in 5m 40s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/f05ce8d72ce44712a952ace5e0ddde3d : SUCCESS in 7m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/18efe09a27594703b2a7ba4bc6b82261 : SUCCESS in 14m 48s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/9ab6959ee7dd4340909073978c4dcb74 : SUCCESS in 11m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/26e068a441a34c8196bee126229f8521 : SUCCESS in 15m 31s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e66fe7ed7fda49d488f307b55cd780f5 : SUCCESS in 29m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f284f9816b6645bca143602c153f2ad1 : SUCCESS in 29m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/bd5e21bcfc4a48e5919c2ba5a6452b55 : SUCCESS in 26m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ee5f80a8dc324a0486af1e3b632d7ccc : SUCCESS in 15m 08s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/8b92979ef391450fb96e0491bcbf6180 : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af11a74759d14b4799689336c88a2cc8 : SUCCESS in 18m 34s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/b65beee3da2542f0b3aa124292a80db6 : SUCCESS in 8m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9919ac9284074b248e2d26d772d7182d : SUCCESS in 10m 12s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e33120a31a9349d3acccff997133cbcd : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38510997718a477ab3869bfb96d9b623 : SUCCESS in 6m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/422822b2f47a49c5bfc8467312cc141d : SUCCESS in 1h 10m 49s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/3565220bb31143788a712f568b4e01d2 : SUCCESS in 1h 36m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2a5028f159bb45f2aae52f01fc2a3ca5 : SUCCESS in 52m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39f1a51049c2844518bc6411279eefc95bacccfc","revisions":{"39f1a51049c2844518bc6411279eefc95bacccfc":{"kind":"REWORK","_number":1,"created":"2020-04-02 11:28:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/716957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/716957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/716957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/716957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/716957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/716957/1"}}},"commit":{"parents":[{"commit":"eb6b078ce373cfef053fc5c1a0ec919541e633a3","subject":"Merge \"Bump pyScss upper constraint to 1.3.6\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb6b078ce373cfef053fc5c1a0ec919541e633a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:28:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 11:28:18.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 1.0.1","message":"update constraint for tosca-parser to new release 1.0.1\n\nChange-Id: Id632090361eb93dfa2f5f4fa735e671d8742ff9c\nmeta:version: 1.0.1\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: ricolin \u003crico.lin.guanyu@gmail.com\u003e\nmeta:release:Commit: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta:release:Change-Id: I899b5cb74e2902c8e5d37a244ebefaa98dbf30e1\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39f1a51049c2844518bc6411279eefc95bacccfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39f1a51049c2844518bc6411279eefc95bacccfc"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
