)]}'
{"id":"openstack%2Frequirements~717032","triplet_id":"openstack%2Frequirements~master~I4983f03c927092b81930baef9480db4aadb8dec5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4983f03c927092b81930baef9480db4aadb8dec5","subject":"update constraint for keystonemiddleware to new release 9.0.0","status":"MERGED","created":"2020-04-02 14:36:25.000000000","updated":"2020-04-02 23:12:38.000000000","submitted":"2020-04-02 23:08:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717032-1585868891334-87f092a5","meta_rev_id":"dd6d9ca7639a1e0e037f5cb2845ec0d7a8bb5a27","_number":717032,"virtual_id_number":717032,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 23:08:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 14:37:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 18:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 18:36:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 14:37:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 18:36:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-02 23:08:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4edc044e3080378c3d64160b2eab85557f7ca8fe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 14:36:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15dde6c812a0323bf46e5631148af87ee2d955c5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 14:37:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"07795202bc941706e39301087c93b28a299e4b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:14:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0faf7187a5d428080c51dd900471748 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241d0da13d67425cb02114a3fadf8890 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2f70fac41c74895a07471bd3b448a63 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b607ebd72f44a6a8fed0732c54dcc34 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2accd3d96ec43619ad82d846093b16b : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96f99bdfd6354b268c2653d5e58d8bbd : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d7f645361974cf28b7b87e2f4e69151 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2ffac36d9ce9471dbb0175e9a244d034 : SUCCESS in 15m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2a75c33739446db90dd041d466fc4f0 : SUCCESS in 15m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5709074c27c44f3db41e3b0feef362db : SUCCESS in 7m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/07fb1e868d1c454fa99b6c99c027b4cd : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff03abad29c04c9fbe0536a890c2dbc2 : SUCCESS in 7m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b7d30d2152784d9f9bd8dedfbdca11d0 : SUCCESS in 13m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d5db1fb6f9d94b4ab755dca6bba82ac1 : SUCCESS in 23m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1653cb44ca2b4de885e302ef3c798f18 : SUCCESS in 12m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/da17927f22594ef08764227782a6fe14 : SUCCESS in 19m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98d74ba513694ab7afa0daf196c9a9b0 : SUCCESS in 17m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e791fb576dde4d23aa879cba3db780ee : SUCCESS in 7m 10s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2de51c200db749cba9ddbcd647b62298 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/891702a833ed41fba59a1e35762079e6 : SUCCESS in 10m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e8d51b713ea14162adbc4d9146afe01d : SUCCESS in 11m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ab9aeeb07c284e22818c2ff1217c3c0f : SUCCESS in 9m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a141cc195844f1ab3ded638208b0e77 : SUCCESS in 51m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/458ec4167ce948bf8754c1df2cab975d : SUCCESS in 1h 06m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6854e4583e714aa0b3c133d5d27913eb : SUCCESS in 1h 24m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/501d2a8563fa4c698e8944a39dbc171f : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2ca9ed1c3fdd77ae4cb6587a01c32665605ec2d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 18:36:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d711bf5df46f280f6afa029986111c20a6952ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:37:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"efa0409d08c979a427a26c53014cb80454ab825a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 23:08:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e9c6fb354e4aa1a4dcba4b290a4431 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1494abbfd04be0af0f1ab35ac4df7c : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cf363ffbbed4a7290c82ca613da6b76 : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a017c1650624d6a94384d86e6386a2a : SUCCESS in 9m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92f9385a219f429c92a39fa7da96f8c6 : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a98e8ed60a5348c5b377b255258d523b : SUCCESS in 5m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e7118c18a8df400487aa491597c1be9b : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0ad8ba511d0545dfaf369a9c1bfb00f0 : SUCCESS in 16m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/abc60a62f7ff40749d11d0f29a4b5b57 : SUCCESS in 7m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/74d7918d4d5e4200a53752ba914acae0 : SUCCESS in 14m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9fc7f258c3f4273b629307e5c6597cb : SUCCESS in 11m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/448608363f22427384731216b3f79920 : SUCCESS in 15m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8357a60bbf29478dadf5531ca2f5d950 : SUCCESS in 27m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f61741dbf43747e3b403798a1bac4269 : SUCCESS in 14m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e8f42a42d47e4341a2f18439a6f94ab8 : SUCCESS in 23m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa7d3a7115264c7eb2739399c6f7cb4e : SUCCESS in 21m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/147c4df79504411095ee5d440f601fb9 : SUCCESS in 8m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4eb4893518c64be880a15ce86ab66fb8 : SUCCESS in 9m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a371747feda4adfa2be66ad82a44185 : SUCCESS in 9m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/285a946631e747f08b2dc4aac6fd785c : SUCCESS in 17m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/abf160215bad4e729befe69bb07565da : SUCCESS in 17m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/790165550d5b47d8ac39c06bb4dce419 : SUCCESS in 50m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab350491bf5841c88ce63357c5851dcc : SUCCESS in 1h 15m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b932efb44d8d4489adb276a4cf375bf8 : SUCCESS in 1h 21m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c3c6a83e5be8409d941e942bed019c43 : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":1},{"id":"aebca96d34b6249ec133cfe78adfcc0beb51cc93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 23:08:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f9a1be5326b9074a790092a2d7a0ca3a067e137b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 23:12:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415c0f63e28648f59cb31b99944acccc : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e41ec44685eebfeb442d5d7c242ead07bbeb207","revisions":{"4e41ec44685eebfeb442d5d7c242ead07bbeb207":{"kind":"REWORK","_number":1,"created":"2020-04-02 14:36:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/717032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/717032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/717032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/717032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/717032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/717032/1"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 14:36:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 14:36:22.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 9.0.0","message":"update constraint for keystonemiddleware to new release 9.0.0\n\nChange-Id: I4983f03c927092b81930baef9480db4aadb8dec5\nmeta:version: 9.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ie1e761515fb7c324d81b73970396a45654fb5043\nmeta:release:Code-Review+1: Kristi Nikolla \u003cknikolla@icloud.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e41ec44685eebfeb442d5d7c242ead07bbeb207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e41ec44685eebfeb442d5d7c242ead07bbeb207"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
