)]}'
{"id":"openstack%2Frequirements~717035","triplet_id":"openstack%2Frequirements~master~Ieb61177ee2b6e31e07159581d3940a87e7a868f6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ieb61177ee2b6e31e07159581d3940a87e7a868f6","subject":"update constraint for keystoneauth1 to new release 4.0.0","status":"MERGED","created":"2020-04-02 14:42:15.000000000","updated":"2020-04-02 23:15:49.000000000","submitted":"2020-04-02 23:08:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717035-1585868893344-6ca8d248","meta_rev_id":"6ec2eca04f130601251fdfef102f6fa8a304f5af","_number":717035,"virtual_id_number":717035,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 23:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-02 18:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-02 18:36:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 18:36:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-02 23:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba7ca1961c1e6628a4701b38c67bf1685c8c1790","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 14:42:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8326b6c221c2aff6e0352f50f7093564a6e02c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b66dfe538d08442b937e31342d8c9a2a : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2e69ea02ca42489a692775e750211b : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb8cde20ba9c4393811a817a8b99dd7a : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a7f62b010504d13a757bbcdc9a44384 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21485576b6224762afbf18a725d108eb : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d5db83d9060419aa8517008d0d17839 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f02f9d348be944e596375c40d9ae5843 : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9de205b8f71446c8957a72afd1aa66f1 : SUCCESS in 7m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70695098792b467394beeb88206dd9dc : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/09662f5e5b4347f1beb64212dc182cd5 : SUCCESS in 5m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/927cb03d999a441ebe4df4f8de55c440 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da82935ec5194ad7bccff9fa9fc9dbcb : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6782451093764c0a9a7102a755f0e829 : SUCCESS in 15m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c1c42473c7884f3181aea54612dd5620 : SUCCESS in 29m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/89320830a2e64bbd95cadbd89e661dab : SUCCESS in 13m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9675fec35d50463d9f52a309de6c137d : SUCCESS in 13m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e9d51b5665642a2b038fc94e8753d1b : SUCCESS in 18m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/48746eb5946643ccbf8d0354aaf79e6f : SUCCESS in 8m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6a303fd8843946baa520e80e5a6685e3 : SUCCESS in 8m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3fbd8d1843e747339130fcab05b91b55 : SUCCESS in 9m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5d72bb8a6cf04e25a32d332c36fb143b : SUCCESS in 9m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0e6ae97629bc402ea4db7624c99ada50 : SUCCESS in 9m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97c048718c83459eb5ca7905afe673c4 : SUCCESS in 50m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55930f52a3b94a6e92aef4ff2704df4a : SUCCESS in 1h 06m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36b6951adaa44d03abf154b147e79d24 : SUCCESS in 1h 27m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ea53bb81509411aa5b096a22e2ffd06 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d973b0ed0a31e6cdd651bdea26d47538fa47fc95","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 18:36:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12dc22a21342756032c5ff6dd0f8292fe3925178","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:37:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8740f699df28abf7ae973ffa3ea941c068acc725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 23:08:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56dceaeed21043b5991a185b820fb86b : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5298508d76a64479a33501ebe783f43c : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f767da77e5454034acc5bd3f44726ddf : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a293af6ee58c4eaab769e8d89ecb692f : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43ad2711ada844d481e45ae1c306ea28 : SUCCESS in 6m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf96602a600f4c389bc37858e1738338 : SUCCESS in 7m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2aaa22d8f4664f399e9e8eb005e6b3c1 : SUCCESS in 13m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f38927eb1fbe4b01a03e0bb5cbc6f8ff : SUCCESS in 10m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/55ed6056f3574452b653a89496bf83e6 : SUCCESS in 9m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5cdd0b068f874e3cb0581a4a5fd91490 : SUCCESS in 14m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a276ef647674f01acab844ac0d0c209 : SUCCESS in 9m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/caab621f730a41149cc19740a09bc8f4 : SUCCESS in 19m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b568d33df2614aeba3d137bbb002d5d2 : SUCCESS in 25m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc6cc34fb46c428b969bec0ed11cf227 : SUCCESS in 15m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bee85b4e79ba4f94bde22ceb9e66a2e3 : SUCCESS in 17m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/363ce6e913184018988d32a05bb7bafc : SUCCESS in 20m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5a1c310918e44df682a68ee4e5518082 : SUCCESS in 8m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c67a23a6089e43e68e55aa3d9d233109 : SUCCESS in 8m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/acb130ba80c64fdab602abbf6c9f9723 : SUCCESS in 9m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d83f96aed0434ee5b79d6957bae4c045 : SUCCESS in 14m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/93c85122b68d4b0dbe6e101d1baa9bdb : SUCCESS in 14m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4fc0f748eef4e4c8faaade5e1a2da00 : SUCCESS in 57m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/56ac56ef8a8248d59c209419e530d409 : SUCCESS in 1h 20m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35d6d260d61c428cae035dc9630c043d : SUCCESS in 1h 27m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/df18a326dba94d66a1448602f71aada7 : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e1c9f472e199bec45911726f13fade414f4aa3e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 23:08:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28c63ba5e8261404e08d50ec821f64953c17b827","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 23:15:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c5d9e2c53d740979fcda7d7d7d1e962 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4dabd81bf79c146e7d3dbb05b044f2d03260d32","revisions":{"e4dabd81bf79c146e7d3dbb05b044f2d03260d32":{"kind":"REWORK","_number":1,"created":"2020-04-02 14:42:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/717035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/717035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/717035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/717035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/717035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/717035/1"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 14:42:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 14:42:12.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 4.0.0","message":"update constraint for keystoneauth1 to new release 4.0.0\n\nChange-Id: Ieb61177ee2b6e31e07159581d3940a87e7a868f6\nmeta:version: 4.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ie49355272ea0f175dd4a0ff5ff078977c9c7ebdd\nmeta:release:Code-Review+1: Kristi Nikolla \u003cknikolla@icloud.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4dabd81bf79c146e7d3dbb05b044f2d03260d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4dabd81bf79c146e7d3dbb05b044f2d03260d32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
