)]}'
{"id":"openstack%2Frequirements~717089","triplet_id":"openstack%2Frequirements~master~I037d8a3e77eceec223f5cfd2abff94e887c0fea1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I037d8a3e77eceec223f5cfd2abff94e887c0fea1","subject":"update constraint for ovsdbapp to new release 1.1.0","status":"MERGED","created":"2020-04-02 18:11:02.000000000","updated":"2020-04-03 04:38:36.000000000","submitted":"2020-04-03 04:37:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717089-1585888627271-73754ddb","meta_rev_id":"ff220149d9d383e8455ee7fce4f09cb3f1618722","_number":717089,"virtual_id_number":717089,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 04:37:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 18:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 22:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 22:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 18:41:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 22:24:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-03 04:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb9ddf0558f3ebaed9060ec6b15ad2b50f3f956e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 18:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06157d556060d7cf31d31b61cb894460e59556b4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 18:41:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc7536b9203534ca1a98bbd90a18ffb245a1ff65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:27:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa797bb60454d288d3fd64becc95a93 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76882b03c4d8418fa5eb04376f918c6d : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bb00b05eebd4cdcb3b143d269f8cb42 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e1e194271fe462096eb16d3a7a8b6e8 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7f17b964dde4550b9e7f2815c464a76 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19a74b50b00e43bd862525484c8cecc5 : SUCCESS in 6m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae91022e3d684a138f22de9fc32533e3 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a4f233e3c33496f90c4ba7af3c0d91e : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8c4af5e10d6c46eb928924e43c1b00ab : SUCCESS in 12m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/faf50a19062e40de836b24dbe7925d2c : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55d15cd199f14edf9596ec98c4429913 : SUCCESS in 12m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9406b66736574ca9a2055bfa8fc5bf0c : SUCCESS in 10m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/864438ff620948ae937c0a9c51768a02 : SUCCESS in 20m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/89a90a16eb2047a48e07db0bbe9e64a7 : SUCCESS in 31m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/990c1088991c4d3ab6bc64d739d173cc : SUCCESS in 14m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2220344ea9e345c38f86b0b323ce15db : SUCCESS in 16m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d99162505223463d824c69182dfc1609 : SUCCESS in 21m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3902dc9d6094c0d9c1611358c6d0005 : SUCCESS in 6m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3587f0f828e2400cad8d2a7c0817a5d6 : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a26bc57bbb1045e094b71e3f4244e5fc : SUCCESS in 8m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/89d2ebc3bdec4a759ac8818a6b461b4b : SUCCESS in 10m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4acf6edf0811451e83e3b11b56f84458 : SUCCESS in 9m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c66aaf17070741d09d516e24746e7d6c : SUCCESS in 55m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebdf6a98a2294b38867ea3e380c46fb9 : SUCCESS in 1h 06m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c28befbfcfe4c7d9aad2529bef2311c : SUCCESS in 1h 35m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/043841d832a442529b3ddaf6f7bd8a4a : FAILURE in 56m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7faf5d38b120497901dd1956bfe8884db754da16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 22:24:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b69e629c2199e9bba5936fd9f3b835c809a92d0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 22:24:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a425ec8464b8aea47f01084e7a950dc4a750ff25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 23:27:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44ce1154382412bbebd5a105426f6d7 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa40e57db58406f8cc76821ee4fc474 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a0105ffcfe84258a4f0ceb6ae54ea8f : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abf81b258bfb483f8b4a26c314617dea : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/440776d5ded84621a869a9ecbcfbec5a : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a543d902deb247a0955eb625bb1b3447 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f89aeaedcb44445b4d852b41d0c7c38 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/281af90d54c74195badcaf876b88b47c : SUCCESS in 8m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/46733a3015e2438eba321958c1cb80b9 : SUCCESS in 10m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4d279a74868d432eaf155fa6459bda3d : SUCCESS in 4m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b6f568a9f58b4d29bf056f9621c08ac4 : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b8aa72f8f2b42578bde1c0a1974595f : SUCCESS in 8m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/08387ecaacb54ea8a8e94445ac75380b : SUCCESS in 33m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a27dd99d26ce4e85b40f22e6be678905 : SUCCESS in 25m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5e446c8cdf68429f8ffdc40855838a0b : SUCCESS in 14m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0c5c7e64f9af4b9fa2ac98c29cdbd41e : SUCCESS in 13m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c84177ca4b6a43b9b186f2c34dfdfd8b : SUCCESS in 18m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0ac41c07cfa642c8b456fe375a379419 : SUCCESS in 5m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3b8467618a714d7984615a90bf061cd2 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6278d761619472484e4a20e36ed71f3 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c1bdf79c226744ccb04b4fc223497842 : SUCCESS in 11m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/32fa8436648346ac9dfd86435c5644d6 : SUCCESS in 9m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04dcca9a9ac940db93b9280c970da5f3 : SUCCESS in 56m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/451cec54b5e44fa88182a60c26123d92 : SUCCESS in 59m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb3fab9466c046c5856c0216b06cf406 : FAILURE in 1h 01m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2892054cd3014f91b1a2a81a0dddb014 : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":1},{"id":"006b11e9405370b91b06ab0d1d21468e206ebe16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-03 00:39:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"592d17312a7863e555e03cb490f72f722f4ae1b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 02:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d55adeef064bbdb9455ca8a93bb192 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8aade1b54c4489b91191abd7784bbd0 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5e36ca797534aac9c6acb171b5d9c73 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cf16917dfeb445f987299e7b56a6a3e : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddb513b6d83e404f8cd90dbf75c09dfa : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad3499a9e61844a58ea130d1fa2b1cb4 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ae4b345671c4c96a95cdaab1fe929d8 : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/16316bd1a1d145678f76859070321a6b : SUCCESS in 9m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2323b4be8b44618a11ee3e156dc3721 : SUCCESS in 13m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7f5cf21f14164ece8592e739a15f7468 : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/91fdce496639414e835bf80bf16e0091 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee98bd140c2b434e8267b7a09caed5a5 : SUCCESS in 9m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a07a25c9fb74cfba0cf9ea60f897ed4 : SUCCESS in 31m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7770828d1cf14058ac68b12a6b483042 : SUCCESS in 23m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5966f7560cba4766a5e8d2343cde25e9 : SUCCESS in 14m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e4acfbdeab944c7389f19639eaa27844 : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c646e2e030f4f9da269ed7b131bad33 : SUCCESS in 19m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e1a191d9794a459a87a80c17cd9a551c : SUCCESS in 5m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/201efae9a64d48a494ec872bebc8fd76 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37aa56642d174031968a4bbfc23b65be : SUCCESS in 8m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/18ca26cf33fa48ce8f5612a9d46b1abc : SUCCESS in 15m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2bf5e37c38324824a3c8be6314c39c63 : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e5f23e3eb5d451bbc1ef03ba557bd98 : SUCCESS in 1h 01m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89002a23a71e4193a654523ca1d3b6d5 : SUCCESS in 1h 05m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/489557f969d34a9e9196b5f187bbc2d7 : SUCCESS in 1h 28m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f8b1eb459f7549d6bddc4ca9c3fc2204 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e7df3332190c9c68edd287c3a43e5ac543cd6cc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 02:10:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"682bfbd8808dbc52835b7fe9a4f57badb7b7d144","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 04:37:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"51120564026a54892c255e69cf5f6be6faab2270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 04:37:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b270c4cd1e6421387b4d7cb2984be64 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2851a81faadc44b89209c27a34f36cda : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e07d7cd116640b1be43037e6ba547c2 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29d4d038c4864663bb9ff529a266ba95 : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85409ae5421747b68fee015f5ec3d965 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab838852f8794edb85414b852f9b6c09 : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a1530af449c4e47b1f5985ae79c8ac8 : SUCCESS in 16m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a36996df129f45f9b2e3d3bf2427d61c : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c75ae195cc14b1ca2499b9662479346 : SUCCESS in 8m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/84096b0a256a4e4787ab0d6d09ca9f41 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d159f7071cc14075bce9b52bd5cf08fd : SUCCESS in 10m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bf165c597996405fa037da28962a4af3 : SUCCESS in 12m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5d3f0e87ea0417083fd284e2a29e27f : SUCCESS in 26m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/415f40af8d334bc3845c4619aa621873 : SUCCESS in 18m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8e2c3404942643038f3623000a4ab889 : SUCCESS in 21m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7c98d0d13a74ad98672d6720adda6fa : SUCCESS in 21m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1652a829e19f4ba9ba8ebb828d4e11e4 : SUCCESS in 5m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2e50acdd57214c0086b8a47bf5dbbed8 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b0accd8aebc41c4aa612755755ace6f : SUCCESS in 8m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/55c22551ee424e01b31e624326d4019c : SUCCESS in 15m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/160f9d36bc47407f95b7f10cfae94b7c : SUCCESS in 14m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f846217e6e984431a7f334ad8f7ad657 : SUCCESS in 50m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/812bed0dce13405e9abb465bfea2084f : SUCCESS in 1h 07m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28911c8f10b44dcbaea63a885806b2d5 : SUCCESS in 1h 35m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/981d7842600d41a39e79445d098dcbfb : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5ca16f10669ebf36a06d85e1af3ee15ef61ede9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 04:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1857a82724e48208c8acdd6a3220609 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b8c7094420f294207c0c775d230064fc49164b3","revisions":{"3b8c7094420f294207c0c775d230064fc49164b3":{"kind":"REWORK","_number":1,"created":"2020-04-02 18:11:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/717089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/717089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/717089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/717089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/717089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/717089/1"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 18:11:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 18:11:00.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.1.0","message":"update constraint for ovsdbapp to new release 1.1.0\n\nChange-Id: I037d8a3e77eceec223f5cfd2abff94e887c0fea1\nmeta:version: 1.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I886aa583a5b38f498596f0b7fd3ac25a58ac75a5\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b8c7094420f294207c0c775d230064fc49164b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b8c7094420f294207c0c775d230064fc49164b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
