)]}'
{"id":"openstack%2Frequirements~717099","triplet_id":"openstack%2Frequirements~master~I379a8029b61bd0827191708944532af90d9cd6bc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I379a8029b61bd0827191708944532af90d9cd6bc","subject":"update constraint for neutron-lib to new release 2.3.0","status":"MERGED","created":"2020-04-02 18:38:47.000000000","updated":"2020-04-03 00:36:12.000000000","submitted":"2020-04-03 00:34:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717099-1585874070414-38ce0ed0","meta_rev_id":"0bffe6c95eafcf61a0d0783b4c0b3b4656690cc7","_number":717099,"virtual_id_number":717099,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 00:34:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 18:40:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 22:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 22:24:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 18:40:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 22:24:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-03 00:34:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdcf2226419bd93e6a06a1d8d86c3131929ad723","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-02 18:38:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcdbae7979b1d2be06c5b85f0479c663300b6709","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 18:40:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5511561131fd4b43c6a36320ed6dd8ae89922ad6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-02 22:24:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0457a3b07c44c55f56122553f769ba76aeeca77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:49:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f0c1bb6e5d4a999e0067a59bdd0a1f : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcdc17da463f4e55bdd43f04cdb5e5f1 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdfc02d0a3894587a0ff32d4630267fb : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/730334fccb5b4b778550d87698a47f4d : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bc2f487191f4ea8b01171494b67f968 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f227895963e41a98b570d7cb9753869 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8ca73563df94c6fb76711a6a98d73a9 : SUCCESS in 9m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ad53fb3c2abd439f8cb3c73fc5c972ae : SUCCESS in 11m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0588ebd5df344c6f9501465a6adbc571 : SUCCESS in 15m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/73383d5e466742cbaa2fdc6d08a1963e : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b26133f4c72141b8b48f11cb2900daaf : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0057af1a385546449e5e040b02286a25 : SUCCESS in 8m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/99cb04061ed54fc0a7910007548528f3 : SUCCESS in 17m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ff3f82b14b7b462ba07c2499d4b033b2 : SUCCESS in 26m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/32143fc78c4f4209a82e8356f50126d4 : SUCCESS in 16m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cf54149caf214dd2af56b6e86e7c28b8 : SUCCESS in 15m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a05628e7a89f411fbd93e2c370a3f1b2 : SUCCESS in 17m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1edee012c2ec4e049cfd7c4d47f503e9 : SUCCESS in 6m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bd0300aad35b422787f066cd0002e847 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/561d1b74737844c0a9975ab94e6234dd : SUCCESS in 7m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8d63554257a44585b6de6c042aafece9 : SUCCESS in 13m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5fabd0d1cf03474e9cb317ed820f708b : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e04120f39a0746d9907c8095e1ec3a4e : SUCCESS in 1h 00m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a326f546b1c04460b62eac38ce3a0a8c : SUCCESS in 1h 13m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97be86cb96af4a2d9af7837fdcf608a6 : SUCCESS in 1h 28m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1373995c1a0f429084041cd6a8401e88 : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":1},{"id":"16919dab229a0e33227f95d2d974d3e028abb588","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:51:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f302d8c0618f65cf71137c26aedf26333e11526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:34:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a59094471e44910ad75c45baa76b23c : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e83f5f0668485f98f17ecb5c4d717c : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e3985706ff9474d97780bb8e3c35cdb : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/694c564063e0430c837da7e8463360dc : SUCCESS in 7m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/122d506bdf4047738abf6a57ebea7053 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7977209f10f44c549ed6ed32c51e5abd : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fe354f11676847e3a971ac9237469357 : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5570d15f74884335800100771add46c9 : SUCCESS in 10m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3a1e34a860df4cb08c20b100bb4429ca : SUCCESS in 7m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d95f56d32ac9442a91b6f51f81cca02f : SUCCESS in 13m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/665f055dfe4344e8ac9b15a591158439 : SUCCESS in 10m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6ff0a9cea4a64056991cb02a5cfb4653 : SUCCESS in 18m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/90dea40de430439abde49cebc5a7beed : SUCCESS in 42m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ea1ccdbb32864972b0047c76dda22c2c : SUCCESS in 14m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/70ebc708fbdb4a99bcbd06ffcf5f713b : SUCCESS in 15m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/591b8a99d2c84c9194ba6a7959715475 : SUCCESS in 18m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d6990b62d46c4923ab79f00784e0b7ca : SUCCESS in 7m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b79b1a52e27949d88a533b2ca0e26ebd : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/723969c39f274526a1dece7e72664684 : SUCCESS in 10m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/56621a9297cb46148cf80a71571104f3 : SUCCESS in 21m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/94688c6f71ea40e7952d0a7beb83a545 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/699939c9458d41f29d81570f4c19f1de : SUCCESS in 53m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/26a182bf6c7f42489dc76b67203ff40b : SUCCESS in 1h 14m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0321c5bc02041abafffb2bebb0e54ec : SUCCESS in 1h 33m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ff3ec4f1751d46ea99b48cb12fd90f46 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5012c60e0e7106b3aa3918bb14fac402dc2315a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:34:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df69bb2d0e048391be888759edc06ecdca42a813","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:36:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221f21d2adf14b07a25316e93ed56243 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"804834edf93adafce001b158534eab500545a7a4","revisions":{"804834edf93adafce001b158534eab500545a7a4":{"kind":"REWORK","_number":1,"created":"2020-04-02 18:38:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/717099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/717099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/717099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/717099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/717099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/717099/1"}}},"commit":{"parents":[{"commit":"75018c0e71d265f77b383af5a50e4dce18155ef8","subject":"Merge \"update constraint for reno to new release 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75018c0e71d265f77b383af5a50e4dce18155ef8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 18:38:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-02 18:38:45.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.3.0","message":"update constraint for neutron-lib to new release 2.3.0\n\nChange-Id: I379a8029b61bd0827191708944532af90d9cd6bc\nmeta:version: 2.3.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Icfe091e3255ab60533c7c5b4f512f1929af74b94\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/804834edf93adafce001b158534eab500545a7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/804834edf93adafce001b158534eab500545a7a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
