)]}'
{"id":"openstack%2Frequirements~717504","triplet_id":"openstack%2Frequirements~master~I7054ac74c253f09a5ae0bf21d14ee22dc75669f5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7054ac74c253f09a5ae0bf21d14ee22dc75669f5","subject":"update constraint for mistral-lib to new release 2.1.0","status":"MERGED","created":"2020-04-04 16:56:56.000000000","updated":"2020-04-04 20:57:03.000000000","submitted":"2020-04-04 20:55:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717504-1586033739580-d5d81bf2","meta_rev_id":"ff7407fa6a6079b89fa8aebcb07af81d172562f2","_number":717504,"virtual_id_number":717504,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 20:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-04 17:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-04 19:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-04 19:17:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-04 17:45:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-04 19:17:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-04 20:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2988c2e8ff50c3b6208d853fff0c8bfc54ece51","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-04 16:56:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd593b7dedc4da0af9f9bf665691a5af4c1612fb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-04 17:45:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5cac7a8d47905c03b8eb81387e56be455494f6e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 19:15:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5893e81422141ada1a8ce4bbea41e1d : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d9f737af0545d5a05410f1a10076a3 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b9e4448898a450788ded6d3563ad487 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25b6298bc96f482da576b865f330e3f0 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1c24765e36b410fbfc62bc05ba73ad0 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c36e2c79bcc415e80383279b2fc6d92 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/564d6c170a55414b9acf94d7b3f10121 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c57ce6ba0a14833b01f98c0f0f16e7f : SUCCESS in 14m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2404b3f93c8f4be7a1587868c85a6e2b : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ebb93170d9594fbc9609d0a0d3ab7f61 : SUCCESS in 5m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/79a18ec3fbc049cf87a74d22e6d0c81b : SUCCESS in 13m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba298a2797fd49b9acc298cc99924457 : SUCCESS in 10m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/67fd216b2418429189341de902845019 : SUCCESS in 25m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce3c724a44774cce9fe9276aff5ef4d4 : SUCCESS in 24m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df3fc58d515b4cadb78f201d4135ea98 : SUCCESS in 11m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/58fbb2e3f79e40208a6aed19831cb3a9 : SUCCESS in 12m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/157c97c35f0843738ab69f6e4fe681d7 : SUCCESS in 16m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/94103f5f92864de4a20ecd416137f5a8 : SUCCESS in 6m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fc47e42c35ff402fb229451fe3f60eb4 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1fa318dac59b4ca4aa9aef0b1fbfd7a5 : SUCCESS in 10m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3c5509e2f06140a0abeccb64a5c8ea89 : SUCCESS in 12m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c738e3aad6c74a95b117f418eb342b6e : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1aff55819ca474fa061c898e0329eda : SUCCESS in 54m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d32c6dc019a24c38928871871164e9dc : SUCCESS in 1h 02m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd8c2c5066d949cd953606479fa7e053 : SUCCESS in 1h 22m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/07e71322df48419bb7bee528abd5b5b2 : SUCCESS in 1h 04m 20s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/testtools/testcase.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/paste/deploy/util.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/taskflow/atom.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_context/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/debtcollector/renames.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/taskflow/atom.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/scss/compiler.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/selenium/webdriver/support/wait.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/netaddr/strategy/eui48.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/scss/cssdefs.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/cinderclient/client.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/prettytable.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/scss/namespace.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/prettytable.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/selenium/webdriver/remote/webdriver.py in the commit\u0027\n  Job cross-horizon-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/selenium/webdriver/remote/webdriver.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/netaddr/strategy/eui64.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/migrate/changeset/__init__.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/mapper.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/migrate/changeset/schema.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/mapper.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_messaging/notify/listener.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/netaddr/strategy/eui48.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/pyudev/monitor.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/prettytable.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/util.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/migrate/changeset/databases/sqlite.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/route.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/route.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/route.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_messaging/_drivers/impl_rabbit.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/prettytable.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/migrate/changeset/databases/sqlite.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/util.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/suds/wsdl.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_messaging/notify/filter.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/subunit/__init__.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/route.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/mapper.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/suds/resolver.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/evaluator.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/prettytable.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/subunit/__init__.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_utils/imageutils.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/subunit/__init__.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/prettytable.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_versionedobjects/fields.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/subunit/__init__.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/mapper.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/routes/route.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/migrate/versioning/script/sql.py in the commit\u0027\n  Job cross-osc-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/docs/lib/python3.6/site-packages/barbicanclient/__init__.py in the commit\u0027\n  Job cross-nova-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_policy/policy.py in the commit\u0027\n  Job cross-nova-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/pymysql/cursors.py in the commit\u0027\n  Job cross-nova-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_policy/policy.py in the commit\u0027\n  Job cross-nova-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/pymysql/cursors.py in the commit\u0027\n  Job cross-nova-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_policy/policy.py in the commit\u0027\n  Job cross-nova-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path nova/api/validation/validators.py in the commit\u0027\n  Job cross-nova-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.7/site-packages/oslo_policy/policy.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_context/context.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_policy/policy.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/pycadf/identifier.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/oslo_policy/policy.py in the commit\u0027\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/suds/resolver.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/cinderclient/client.py\n  Comments left for invalid file .tox/docs/lib/python3.6/site-packages/barbicanclient/__init__.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_policy/policy.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/evaluator.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/suds/wsdl.py\n  Comments left for invalid file nova/api/validation/validators.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/pymysql/cursors.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_policy/policy.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/oslo_context/context.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/unittest2/compatibility.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/enginefacade.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/pycadf/identifier.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/pyudev/monitor.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/eventlet/patcher.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/netaddr/strategy/eui48.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/subunit/__init__.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_messaging/notify/filter.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/routes/mapper.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/routes/util.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/prettytable.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/neutron_lib/context.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/rfc3986/_mixin.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/migrate/versioning/script/sql.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/migrate/changeset/databases/sqlite.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/webob/acceptparse.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/migrate/changeset/__init__.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/scss/namespace.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/sqlalchemy/orm/relationships.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_messaging/notify/listener.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/scss/compiler.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/rfc3986/_mixin.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/paste/deploy/util.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/selenium/webdriver/support/wait.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/debtcollector/renames.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_versionedobjects/fields.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/testtools/testcase.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/pymysql/cursors.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/selenium/webdriver/remote/webdriver.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/routes/route.py\n  Comments left for invalid file .tox/py37/lib/python3.7/site-packages/taskflow/atom.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/migrate/changeset/schema.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/netaddr/strategy/eui64.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_messaging/_drivers/impl_rabbit.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/scss/cssdefs.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_db/sqlalchemy/exc_filters.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_utils/imageutils.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/oslo_context/context.py","accounts_in_message":[],"_revision_number":1},{"id":"144a5bed3aa34dbe35e9acdec7574c7bfdd91444","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-04 19:17:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a30eac336746bcf6fc0af61f815e12ed2db28f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 19:17:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"915dc1e454a76d6e5bfcadd683a9a821c5998894","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 20:55:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c229e09659440f3817d3e460371c893 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/714bfd7684914eb1842f43f4cc3d65bd : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0dcc29d22554bc5b11434a7d87f15c3 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/625326ce0d914788bc23b4eedf11fdd6 : SUCCESS in 8m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9d70b75bb67458eb6046adc28ae01e6 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46f2cdf8071b401d9474427a8b52378a : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5dc96926e304fc78419244763af95e4 : SUCCESS in 17m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/42e7497daa1b4438ab5058571a672177 : SUCCESS in 20m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1fedc5bd43a94beb8f84f739b79e075f : SUCCESS in 8m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d16235dada6c4f6fa1363305d75e661d : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/061880bdfd814c68b8ad4f74f786f286 : SUCCESS in 6m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/88081db5e8a841ada40254b9b8141996 : SUCCESS in 48m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d8f9b0d5f8ac4a209c3049c594d137b4 : SUCCESS in 38m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/38d8594068ae4236a6d4e5de42197ec2 : SUCCESS in 11m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e9d5c1991a464afb9b919ff5fb5f839c : SUCCESS in 33m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/598e538286334060bf3b2b0d3eaf59a1 : SUCCESS in 29m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/01fd98bb5ebe4ab7a4195654107a929f : SUCCESS in 5m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2bc9efcb8a484f70940c82c4ede44c82 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce75287fadc64bfd98bdbbae9dc0f8d5 : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fc252442c92a4bfa90282bf7001f9519 : SUCCESS in 8m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c2453f32d4554eea9c0e5447d07f1622 : SUCCESS in 9m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba307be03be142c98e09b11304b178e0 : SUCCESS in 53m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d708a566f7a14574b90157d68403ec0b : SUCCESS in 1h 08m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f358634cff14f74a122d564418af7e0 : SUCCESS in 1h 35m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c334d4015e59488eb0dc25319488ca43 : SUCCESS in 1h 02m 25s\n\nWarning:\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-nova-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path nova/api/validation/validators.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file nova/api/validation/validators.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":1},{"id":"becc685077471e74f5090b7e3884a435e366134a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 20:55:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"69fd43efe6d32ee2d09d5942ef1dd9d6573d57c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 20:57:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c780f39edf34a63895e2ded790508da : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62d4c19a5e740fe122a946fc2cf7a6d38a599c7d","revisions":{"62d4c19a5e740fe122a946fc2cf7a6d38a599c7d":{"kind":"REWORK","_number":1,"created":"2020-04-04 16:56:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/717504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/717504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/717504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/717504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/717504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/717504/1"}}},"commit":{"parents":[{"commit":"b0b27ce07f22678fa14fde32b78fa6de5d6b8dfa","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0b27ce07f22678fa14fde32b78fa6de5d6b8dfa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-04 16:56:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-04 16:56:54.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 2.1.0","message":"update constraint for mistral-lib to new release 2.1.0\n\nChange-Id: I7054ac74c253f09a5ae0bf21d14ee22dc75669f5\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Commit: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Change-Id: I4a826f4e78ce509e2e503c69f019c97c4cf1cb07\nmeta:release:Code-Review+1: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62d4c19a5e740fe122a946fc2cf7a6d38a599c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62d4c19a5e740fe122a946fc2cf7a6d38a599c7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
