)]}'
{"id":"openstack%2Frequirements~717772","triplet_id":"openstack%2Frequirements~master~I8cf2a0932701b7023eef22b5cb2b6d3088891389","project":"openstack/requirements","branch":"master","topic":"sphinx3","hashtags":[],"change_id":"I8cf2a0932701b7023eef22b5cb2b6d3088891389","subject":"Lower Sphinx constraint","status":"MERGED","created":"2020-04-06 14:53:51.000000000","updated":"2020-04-07 05:14:07.000000000","submitted":"2020-04-07 01:47:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717772-1586224028604-87f092a5","meta_rev_id":"654ac2dbb6b5d32512332122b95aac3a0795a6b8","_number":717772,"virtual_id_number":717772,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 01:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-06 19:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-04-06 16:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-04-06 19:25:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-04-06 16:06:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-06 19:24:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2020-04-06 19:23:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-06 19:33:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":1,"date":"2020-04-06 16:47:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-06 19:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-06 16:06:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-06 16:19:58.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-04-06 16:47:04.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2020-04-06 19:23:55.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-06 19:24:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-06 19:25:44.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-04-06 19:33:27.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-04-06 19:34:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-07 01:47:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57dea326abebbdf38be83dbfb6fbdac6f9f573a4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-06 14:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a6dbaff2bbdd528c196065c764e392c44b2b1ae","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-06 16:06:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4ae0b12a6cc9917ca33c1ceac0f05fc091475131","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-04-06 16:19:58.000000000","message":"Patch Set 1: Code-Review+1\n\nAgree with Sean here.  It\u0027s feature freeze this week, highest priority is to get the gate un-stuck.","accounts_in_message":[],"_revision_number":1},{"id":"10af32082afac71e947df40a2be6d0109cf727ea","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2020-04-06 16:47:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6e72b49d4f2a928e5239bd8739aa0c854abc8b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 17:53:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da6a2114dfd64137a4c1252523ac4981 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f33b37b620643588f98511a98359287 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b317a2c8a7b34316b3db57bcfbb7d063 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6aa2d0b7be0422789014f481470bdb5 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e3f0f243844760b0b3fa600f2b349d : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf861d16790545e6b25b7666ffdd13fd : SUCCESS in 7m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68014270918f418f9adc3883e50f8d75 : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d00b8ed7b1324a3da37bfdd8aa6a4084 : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2d3539c00e994b3b9d037a3979755ab0 : SUCCESS in 9m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/347d62f5ff3c401db2643936430128d4 : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d91f8636fbaa47adbcbbfe5ef558e307 : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6531f6a1ed246539d1ac4562ba93b0b : SUCCESS in 9m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dd8b0315cfca4c68bd2b5f77bc96e099 : SUCCESS in 19m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d4e8b2b12bbb456bbba6721684f3bcf0 : SUCCESS in 29m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b60e4001c4924aaf9750fda1e4b72bc1 : SUCCESS in 13m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f6d86f9165ad45d8a5abdd8f5274725c : SUCCESS in 18m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1156e7495b804a05891c5983fb231b07 : SUCCESS in 18m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/123a3b85440e4a41bcd44d4cf3b46a8c : SUCCESS in 5m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3001dd7a71ac4bcd828de0bd38154f4a : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18e8732087fa472894b249c52ad49f1f : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/62fe4c6a06d94986b95ab31ee5301567 : SUCCESS in 12m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d0a1397b3bfc49dba75408d17001dcae : SUCCESS in 15m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1de109c2582473a864789996b98b3ae : SUCCESS in 54m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c19dba377e144d67959ab2eecf19b612 : FAILURE in 1h 06m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4924df68e15941e1a80ff9ee9feb2b4f : SUCCESS in 1h 32m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ffd4fe5ba3124392866b87cab8d59547 : SUCCESS in 1h 12m 47s\n\nWarning:\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-nova-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path nova/api/validation/validators.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file nova/api/validation/validators.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":1},{"id":"a8ac1235a58a619748537e5e9a012c346d37cf84","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-06 17:55:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1cbe5b587f86fdd7f601b2c0496b10de72d55176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 19:10:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db8e1c1450c4245888d88e5eb4399cb : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/615c780982e44ac788fee1cc09db1536 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d03a6231b86044a0ba281ae27ae14005 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/982b65f453324f198ef0b3c4643d26fe : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c96b040ed694c3791893a27f50c494f : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a8d22a9a4d34e24bc215594c4c7c4e3 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/363c164c7e054ca5adf6f3711147e7a3 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3932f2d42b4649558d60624f7dc99582 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/96102cf5aeca45dfacf43f1c5bc52bbf : SUCCESS in 11m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e15709d92e37429d9fecce46b8e4f6fd : SUCCESS in 6m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3f22261e8ed043ca8992bdf0da9b701a : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5971def6f82c4cfda5c7eb07191416a9 : SUCCESS in 8m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/43d40dfee46d49cc856a4130dcf246c2 : SUCCESS in 13m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a79139464e1d4e05ab89a9c2adb37498 : SUCCESS in 32m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e1ed3d62c579418f8bd4dbe35e4aa6fd : SUCCESS in 13m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/db6b24ba86144b4999e9a17433ca298e : SUCCESS in 15m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2752631e16f24513aa1ca4502f484efc : SUCCESS in 17m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/27ddfa13b60a4b46934e9b9ee1eacbbc : SUCCESS in 4m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7b8f5665ca954928a83298c7a38f76f7 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8aa0dd0c0ba140bfb2146facfc26aa97 : SUCCESS in 6m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9d2a488918bf451d8a889d3148c620bf : SUCCESS in 11m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2ba95daed7964503986034834f48d16c : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7603e70c2a30433bbccbf327a693d968 : SUCCESS in 46m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/019dd6eefcc9482fb58e6497e6301984 : SUCCESS in 1h 07m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77b1bdcb432d4443982350f7fe821703 : SUCCESS in 1h 13m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2e1b46e9f2fe48e4aa92dec09b23d045 : SUCCESS in 1h 03m 45s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-nova-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path nova/api/validation/validators.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file nova/api/validation/validators.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":1},{"id":"3690ab048c0b003a8584b84fb55ab40273de0652","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-06 19:18:39.000000000","message":"Patch Set 1:\n\nif we need to roll this back, a masking of the known bad version should be included in the review.  please add !\u003d values to global-requirements.txt","accounts_in_message":[],"_revision_number":1},{"id":"6d9142f9cc3283333f666548b3475af005de5a6a","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-06 19:23:55.000000000","message":"Patch Set 1: Code-Review+1\n\nmanila too","accounts_in_message":[],"_revision_number":1},{"id":"c3e87cb8301f6e61b0e8d32832267aaa4f3adef1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-06 19:24:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"584e4706ccb4a201c95392d7c02d261d7b5623b0","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-06 19:25:44.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks Sean!","accounts_in_message":[],"_revision_number":1},{"id":"f03e7a7b0366bc37d5ff921487ae0172a6a7222a","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-06 19:33:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0812a409eff3a1ef06e630fc53ddfe1d53e0e6ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-06 19:34:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ncan be follow-on since it breaks things now","accounts_in_message":[],"_revision_number":1},{"id":"020197efdf85532b321c38b78c873aeb0ad311bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 19:35:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc93d120fa4d3afd2ec067bca76d7c8ff582ddcd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-06 21:37:17.000000000","message":"Abandoned\n\nShort circuiting transient gate failure.","accounts_in_message":[],"_revision_number":1},{"id":"857610f7ac3a40209dbd2222ed4da0b830dad411","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-06 21:37:28.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"7f164cbd50f7dfa9a83143663095ff49a6c797b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 23:02:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a65e955cb1f426597c5344dc25110d5 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d885f2d17ab64859b69f3b002bcdcc88 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55c9323d188f48ecaeb435af0f7f7ba4 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a45beaa70fb4ae29ee8f1f6c51b6a2f : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/644726b9f5694511a8421b028b7a489d : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc636acd28d747749f79125f28dd3bbd : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c72b5f8d547142faae39d6e06e549c80 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a0373ea660d34ff88432ebd1f634fada : SUCCESS in 10m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/567847da57c6417bac8cf125a9e6ca60 : SUCCESS in 10m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1d236ba808324cfb98db9a168ded58ee : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3dc0a43e41e045c095f42eed7b44607e : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc9b773e17ec445db6dc2ed673a99c99 : SUCCESS in 8m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e2e46de95dc4422bbf8122d2b17304f : SUCCESS in 13m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d67585390b3a4eb2b1d0e32b2cbdf892 : SUCCESS in 26m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2fb0e7e5fde445598c22783c8df3a900 : SUCCESS in 14m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/490558b8b9d04ba0a438e1133321ec4f : SUCCESS in 13m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/139b9e336c7f4326895b6f8049600a74 : SUCCESS in 19m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/abe96ea2a6374142a0ee2e863d830086 : SUCCESS in 4m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1d4bbe3f341d4311b89b0db9630ce043 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74776208601444b4b06a5226be2a2b5f : SUCCESS in 6m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8c0ad4883fac469ebc847d78f7d0aec3 : SUCCESS in 9m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1bbec594912d4231811fdfce037e6c0b : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a963ab98936a49c385eb06461ef39023 : SUCCESS in 52m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b518a4c3f844626a09dde5876dc0ac4 : SUCCESS in 1h 14m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cb9913d7e5147b394a969068c1f99f6 : SUCCESS in 1h 20m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f552451e25c4979a81b9baf1974f151 : SUCCESS in 1h 15m 14s\n\nWarning:\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"7ae1c34df4bafc5bc0d0fa44977f54ca2226bff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 23:02:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a707e220a42e5a6afd300d0ff153cec26f592901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 01:47:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec417371f46448a6b2951df3d6695ec1 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bde02548cad46c2a8d072ed6bca9dcf : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd8fc563235246ad9ba484ff25e60624 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b38c150428b141cfad5851b0dd041fb8 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76420c1aa9024eeabfa5156a034f3ea8 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16e23f1f0db2486bad96a32840100f31 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/90a20fec1d2b43708f9d1e52282e5e4a : SUCCESS in 8m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/93e9aeb5f1d64afc9de14a37d6c01e8d : SUCCESS in 8m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/70972a3091c0448f87fdbbd308ca7c35 : SUCCESS in 4m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d3086a46e6b140a4914958f52d96888c : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bebfa1dab67a43f89adb988c98fda257 : SUCCESS in 7m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8a72cb8f3f44bb0bf1a9adfd4062d4d : SUCCESS in 14m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/13c4ff4ac9b64c978b431841bba4c5bb : SUCCESS in 26m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/961722d607a6472caec9c6da460c4a42 : SUCCESS in 14m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a766f42ce3cd4e53a0c59327c2a877c5 : SUCCESS in 17m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/808682c0f153434eb6fa6e3aa170d4fb : SUCCESS in 18m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/62a0fdc4571049f59435207e276b98a0 : SUCCESS in 4m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/10d7cfd220f64a3fb0f689a64257d8aa : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/031489daafc54596aa579bebb033f6d4 : SUCCESS in 5m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/75d5e9c23ce84a98997fd52e74123724 : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0bdcf92cdbd5459f8d5501a3e0244ee4 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28f39b459fbe407d9d20c43ad5dfb650 : SUCCESS in 57m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a10f6741e70462cabece957b06293eb : SUCCESS in 1h 15m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9c20110d3b7476e9f661f927fd57380 : SUCCESS in 1h 58m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0e916f9a7e194144a29b4d13b07b24f1 : SUCCESS in 1h 09m 52s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"3077599a230fc167cab0026ae0a15427b97efed8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 01:47:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"36df908f3f01a5db2648740cef66f1966d8f03b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 01:48:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e90ba70c63d042938fbcff2e2028a897 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d0d6053b517a27ea62fbc40b4bff47b9a0b8c58c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 05:14:07.000000000","message":"Patch Set 1:\n\n\u003e can be follow-on since it breaks things now\n\nhttps://review.opendev.org/717949","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2100c093a281f7322e3d3d1da803a11c8db0abca","revisions":{"2100c093a281f7322e3d3d1da803a11c8db0abca":{"kind":"REWORK","_number":1,"created":"2020-04-06 14:53:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/72/717772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/717772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/717772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/717772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/717772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/717772/1"}}},"commit":{"parents":[{"commit":"b59db53dc527049908e972948667f2a73704f3d5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b59db53dc527049908e972948667f2a73704f3d5"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-06 14:52:25.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-06 14:52:25.000000000","tz":-300},"subject":"Lower Sphinx constraint","message":"Lower Sphinx constraint\n\nSphinx 3.0.0 has introduced failures in several repos [0]. Until we can\nidentify the cause/fix for these failures, and given the point in the\nrelease cycle, it might be best to keep this limited until we have some\nmore time to understand the issue.\n\n[0] https://github.com/sphinx-doc/sphinx/issues/7422\n\nChange-Id: I8cf2a0932701b7023eef22b5cb2b6d3088891389\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2100c093a281f7322e3d3d1da803a11c8db0abca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2100c093a281f7322e3d3d1da803a11c8db0abca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
