)]}'
{"id":"openstack%2Frequirements~717949","triplet_id":"openstack%2Frequirements~master~I79d232348a51b95ffac67aa07b46c8da71c3ac0a","project":"openstack/requirements","branch":"master","topic":"sphinx-3.0.0","hashtags":[],"change_id":"I79d232348a51b95ffac67aa07b46c8da71c3ac0a","subject":"Mask Sphinx 3.0.0","status":"MERGED","created":"2020-04-07 05:13:53.000000000","updated":"2020-04-07 20:07:17.000000000","submitted":"2020-04-07 20:04:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717949-1586289857830-4a059950","meta_rev_id":"cbbf404fda3bfb0aeac6e06247551f3ee8fb5817","_number":717949,"virtual_id_number":717949,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-04-07 09:55:24.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 20:04:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-07 11:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-07 16:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2020-04-07 12:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-07 16:04:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-04-07 07:22:25.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 07:22:25.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-04-07 11:19:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-07 12:27:06.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-07 16:04:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-07 20:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d932936529791595c3a9337ffedf2293325bcf7c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 05:13:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1feea860f8c7bff7d8be2b61686d1c70ec6fb9a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-07 06:48:20.000000000","message":"Patch Set 1:\n\niirc johnsom was working on getting a new release for wsme (from my quick read that was the source of the issue).","accounts_in_message":[],"_revision_number":1},{"id":"6cb10a41b3b68048b3abd92e688b7dc3f20d49bc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 06:54:07.000000000","message":"Patch Set 1:\n\nwsme might be the culprit in some cases but not in all - AFAIK.\n\nBut let\u0027s see what he comes up with...","accounts_in_message":[],"_revision_number":1},{"id":"c44d43ee3d828faca6e10977b1713bc44ce2b4fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 06:57:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998adeaacfa2476e8b27009f01026c1e : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dcf5a9f6ace4acb8f1f344f3e41f584 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f4c650109414c448ee074886da44e9c : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/847650a1daa94fc3a4ff823e00b86f30 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11be4fbd1de94784b1f5d936cd6cccd7 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/197e2211910547858274d8b0c7b2e07e : SUCCESS in 3m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/468f21527a6142cd80ec94b1286055ae : SUCCESS in 54m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9e6bd1e9a4a4f1ea9c04c79e5c6f897 : SUCCESS in 1h 41m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10846ae152084d8094f5853ab4bfad11 : SUCCESS in 1h 28m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/919aafa34e434fce84cc8fba17358657 : FAILURE in 1h 11m 33s","accounts_in_message":[],"_revision_number":1},{"id":"deb1289370add5657d49ec054656dca9f3f598ce","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 07:17:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de49be6a5d409a40b0c99a73033366ad30532498","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 07:38:28.000000000","message":"Patch Set 1:\n\nYes, wsme is one problem - but there are more. cinderlib was hit and does not use wsme.\n\nMerging this will allow to block the release for now like in https://review.opendev.org/#/c/717962","accounts_in_message":[],"_revision_number":1},{"id":"e6455fca5d7cbf648c5a4ac8b2cae51093ee276a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 09:34:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4187c5d4234b1cbef8cf0a0894f585 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7441d35a8a143169571770bc23a0de0 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b6dc647f3594eb4bc3c82c67a64e197 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9195b571d93a4a729d615b8083276c28 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8169e6db6e4c4f409c2c298ae4fa7678 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82a63eb22ae645b3b04a35cee4511b58 : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bb46e6b7f2b482ca480ab184ae269d4 : SUCCESS in 54m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7677d067cbe4d9aa63b48cc764ecdc9 : FAILURE in 1h 18m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd40c35f20ce46cebe50c8a822504893 : TIMED_OUT in 2h 09m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3f4f2ffe300b4737aeb772f0c8f16aef : SUCCESS in 1h 15m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2d96ec85299bc65289b02d0b67b4f36b7f21db4e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 09:55:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6110c62bab5a37e9981912335edf75bfbbcf396c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 11:19:34.000000000","message":"Patch Set 1: Code-Review+2\n\nI8cf2a0932701b7023eef22b5cb2b6d3088891389 was meant as a stop gap until we were able to determine if the sphinx build was bad, or if we actually just needed to make some adaptations for the new major release. Based on the number of comments both with OpenStack and from outside OpenStack on https://github.com/sphinx-doc/sphinx/issues/7422, it is looking like 3.0.0 will definitely need some updates. So we should be good to block this release for now.","accounts_in_message":[],"_revision_number":1},{"id":"bfe21ee25345a3491107d1faeec602a8d72e3bcd","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-07 12:27:06.000000000","message":"Patch Set 1: Code-Review+1\n\nmanila doesn\u0027t use wsme and had problems with sphinx 3.0.0 and oslo policy_gen","accounts_in_message":[],"_revision_number":1},{"id":"15fecc4b6ea2bfe7fdfc90032cbb13b2430c04c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 16:01:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7db5d8a64847a1a54d90bbc7bb30e3 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e98d4955ddb48cd812b035e927970e9 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee77a688c9aa4064869d8809df771942 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8f81c5f26124aa395badc79ba074dfb : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/291a41be1e2f49249c655437f59048ab : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08c8fe9ab0124eb6aba0c0880a96d14c : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/754236a963844a3d85ea30b15ef96ebc : SUCCESS in 55m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df718be9500742ba95df23e2408320e3 : SUCCESS in 1h 24m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cf9562780d844da8694caa577b344bf : SUCCESS in 1h 28m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/51d187d57c024f9b8da2903f6708087d : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7de8d357f6c1b0de8ff3d0a8baa9e8cda220a9e6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-07 16:04:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c21c7444bff02b7915f9b460da008bf1aa873301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:41:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0006e9cae2de8aad62915b81501200336cb4d6bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:04:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2953fd342158ec1ed6ac7827ce84dc9b104942f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:04:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dded71be160b4865ba01c62839b5eaf0 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a35856899f402a92fb433095c7d6e3 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bad4bc167884f988e39c668418d46cf : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3f2feffe8d4470ca9a3e3430d35e03a : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e0c930744614edc9306a86b9d78267e : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/543316ed8357480b96954f376f120129 : SUCCESS in 55m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d753508c2e4748678e23edc6ca454011 : SUCCESS in 1h 17m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b725f1541a548e6925717bdc8021a71 : SUCCESS in 1h 27m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d6c02e06725e40ffa2290a8b63c30044 : SUCCESS in 1h 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c8d56db694edd07a2b568514d5c6ce9bcf2b93c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:07:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c4f02a89114d57acc947218ab3259f : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"278058ac80e4f342d2b7f49bce1b9df20f4dc291","revisions":{"278058ac80e4f342d2b7f49bce1b9df20f4dc291":{"kind":"REWORK","_number":1,"created":"2020-04-07 05:13:53.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/717949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/717949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/717949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/717949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/717949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/717949/1"}}},"commit":{"parents":[{"commit":"c75ab1f622598e15d1b51a8d0801384d202630a2","subject":"Merge \"Lower Sphinx constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c75ab1f622598e15d1b51a8d0801384d202630a2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-07 05:12:57.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-07 05:12:57.000000000","tz":120},"subject":"Mask Sphinx 3.0.0","message":"Mask Sphinx 3.0.0\n\nChange I8cf2a0932701b7023eef22b5cb2b6d3088891389 removes sphinx 3.0.0\nsince it breaks docs builds, add the missing mask entry for\nglobal-requirements.txt.\n\nChange-Id: I79d232348a51b95ffac67aa07b46c8da71c3ac0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/278058ac80e4f342d2b7f49bce1b9df20f4dc291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/278058ac80e4f342d2b7f49bce1b9df20f4dc291"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
