)]}'
{"id":"openstack%2Frequirements~718019","triplet_id":"openstack%2Frequirements~master~Ia5002481fa97835f68051dedd95bfb781d9a30e5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia5002481fa97835f68051dedd95bfb781d9a30e5","subject":"update constraint for debtcollector to new release 2.0.1","status":"MERGED","created":"2020-04-07 10:15:04.000000000","updated":"2020-04-07 20:04:21.000000000","submitted":"2020-04-07 20:04:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718019-1586289861465-25f1bb97","meta_rev_id":"c1ca7f1486441192fea19c8668566119c1d825c9","_number":718019,"virtual_id_number":718019,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 20:04:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-07 11:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-07 13:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-07 13:19:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 11:03:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-07 13:19:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-07 20:04:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7125f95c4066eb577f20924cffcac1691804f482","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-07 10:15:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b63eee9187e07cc8806238b927c766e1761c9d5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 11:03:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"957a9450d73ca354164b32245c4cab1987edd599","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-07 13:19:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de027c60c51297d769bbd91474239261f710944f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 16:37:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7362140e79e42edb83a7bd17a17c7ea : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd51e8fbe312487cb16f4866c9964292 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cee9e4b7b61430bb08487918bdeabc4 : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/762ef40ed2aa4721b70336cd648690cb : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccecb2b6f9f44bcb88b87a237e810499 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00c2a66168d640679897eb5b88139f98 : SUCCESS in 7m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9be78868d074363abb353dfab046c10 : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd8ebe8d966f4c909fc78c978bfec4b7 : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b6b5aaa3339c441e973d8b7f9792b8e9 : SUCCESS in 14m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ecaf53143dc544f18d1982cbf3766b87 : SUCCESS in 6m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/599195cd9b2744eabf2b37fd40ac04b7 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/569b837577f54d5fb496a835f3fc74ff : SUCCESS in 9m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/879ed0ce3e48485c8dbc8e937ac6b224 : SUCCESS in 27m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1f989a735cd24d9f933311cdd5b0eb6f : SUCCESS in 25m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2faf5aa2c6364173a90169d20ec46f0f : SUCCESS in 17m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5dd3dcef657145479651315ab6f01e50 : SUCCESS in 17m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ee90d4c207540289acea25a478dfa30 : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c15c690da9e54a0790c320b0e1ab9fc3 : SUCCESS in 7m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6114122579254f62a9a7f200442ffc22 : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aee002cf9f354980956859ff8ea83767 : SUCCESS in 8m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fceea8eeb24a4f9f902c15c11c1d4fe3 : SUCCESS in 12m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3712369d436d405d8b54b12eb2eae83b : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3441cfc8975349cfba960577b92275ae : SUCCESS in 1h 01m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea7c0ac0df2e48909bce14ee65452f10 : SUCCESS in 1h 22m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76892351e13a4c408121dcb4ff1aaa88 : SUCCESS in 1h 26m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1c9e96e4ebc04ab4b70e11e0ae903638 : SUCCESS in 1h 17m 39s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"6d0d1a1a3abca25883be45e310ae4533c5d4116c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:47:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af4a35526e118cba41d7a44d50a5a122c0c185e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:04:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/036344942bac4205a1e827da8893a7be : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d28ff89d5084435499afe2b48533c536 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/726c9001a7804ea1ac0dcaa262de4e01 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/513eb68d11ea40e08603b782fb747ee6 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80984f7c87604b70a8c3b431a31b5405 : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7bcb1044d3a342d59058b3ec9ef70a11 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96bb1d3ea2ea4e988a5bc33c2c7bf2d3 : SUCCESS in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/75acad9342bf40db8d3d6188c20cc75e : SUCCESS in 12m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ca040ee9d4cd4774ac623612ac9518d9 : SUCCESS in 7m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/34a639227de247e390c5cc02544c96ba : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd037fc30d54400686a89870c38a4ee5 : SUCCESS in 11m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2db987bc9bd54e18bbfffebae45dc704 : SUCCESS in 14m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7deb8b28d5ac4ead915276b10fb6e7cc : SUCCESS in 26m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac37f7b833b24c1f8d200e1237a12a5a : SUCCESS in 14m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8d421c228b764af88f751832ba3878ee : SUCCESS in 16m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb3f4320147044d9a74ebee284c85fcc : SUCCESS in 17m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3e236fdf1aeb42a985379fbeb5ee3e57 : SUCCESS in 7m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cbd92ab6f6d34c95b0f81c9f1fb663f7 : SUCCESS in 9m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/324cba213c154e6e8917687a015da7cf : SUCCESS in 10m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6920e0c4da714530b08b3b54831aa14a : SUCCESS in 13m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/271a9ce2c01f484cacd47e9075651b40 : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ce9a62e74434287aff628ebc31cd973 : SUCCESS in 55m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d840e081f9da46bda2bb15ec5438b898 : SUCCESS in 1h 20m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab6a24b7246a4e54a1aaab781f49e2ab : SUCCESS in 1h 16m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f6a3a701e9394478a61630b6ec41a465 : SUCCESS in 1h 07m 43s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"0b28668a9c436eaec314c0c5cbff55c52d85be32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:04:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72f3c04863ae4ebfc8e4000005624fd4b6f4459c","revisions":{"72f3c04863ae4ebfc8e4000005624fd4b6f4459c":{"kind":"REWORK","_number":1,"created":"2020-04-07 10:15:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/718019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/718019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/718019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/718019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/718019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/718019/1"}}},"commit":{"parents":[{"commit":"c75ab1f622598e15d1b51a8d0801384d202630a2","subject":"Merge \"Lower Sphinx constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c75ab1f622598e15d1b51a8d0801384d202630a2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-07 10:15:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-07 10:15:01.000000000","tz":0},"subject":"update constraint for debtcollector to new release 2.0.1","message":"update constraint for debtcollector to new release 2.0.1\n\nChange-Id: Ia5002481fa97835f68051dedd95bfb781d9a30e5\nmeta:version: 2.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ica9d5824b33e3f244d4bff64d6f5d101f5b371fa\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72f3c04863ae4ebfc8e4000005624fd4b6f4459c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72f3c04863ae4ebfc8e4000005624fd4b6f4459c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
