)]}'
{"id":"openstack%2Frequirements~718038","triplet_id":"openstack%2Frequirements~master~I909abcd5a8b8cd52c1df1953533ecacb6e493e95","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I909abcd5a8b8cd52c1df1953533ecacb6e493e95","subject":"update constraint for oslo.serialization to new release 3.1.1","status":"MERGED","created":"2020-04-07 10:20:17.000000000","updated":"2020-04-08 19:53:36.000000000","submitted":"2020-04-08 19:32:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718038-1586374370235-02fefbb6","meta_rev_id":"b7dac96d068d6bc8bfa660e6583f0c3737420463","_number":718038,"virtual_id_number":718038,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-04-08 09:16:01.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 19:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-07 11:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-07 13:11:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-07 13:11:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 11:05:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-07 13:11:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-08 09:16:01.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-08 19:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bd1ca46a9b7f72f8df096d0fcaff74654714b8d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-07 10:20:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93f5f431da1c347c82c781b75e823beebf1414ae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 11:05:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22c9c12d5ab8d3764c20d988af17c088d1e51333","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-07 13:11:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"280dfa779e81f9aeb12133c76ac5f5ea2a943756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 18:18:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d47157628c4796842db7cf43a41263 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1cc2a4b139c412ea40c2b83bd1045b0 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b56832c1847d4c35b93cd4e5f4083dea : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e19c769ebc14273b82bc9ebc84b0688 : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/946acb782ee04ccdb3065706abf700d8 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f80d7896fa554a0d8a86409b4a1221c0 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2af3ecf4c6254d1b831bc4de278f29e8 : SUCCESS in 7m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b26544dc504a4b888c4fdf1bd78f407d : SUCCESS in 13m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9bf7aa73f9bd476c97edd0ba83296648 : SUCCESS in 10m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d1ca7e66e6514dad94d9b76880e825ba : SUCCESS in 7m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2815fa1fb20b483fa89d59b0ad007468 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c99c1ca4d694bb9af517835166d2a82 : SUCCESS in 13m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/02b1da0da8b947a2876af0fe49ce7a58 : SUCCESS in 16m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8900598f18ff45438872f1bc26612506 : SUCCESS in 29m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/758ade41e1bd49f19f1795c46b58267a : SUCCESS in 16m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c27e242e103440358757537f43f605c5 : SUCCESS in 18m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf331f3072424b5cbd1130dcb2a16b4f : SUCCESS in 20m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/578f1454112e47a58bb7cf5f316fa6b2 : SUCCESS in 6m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/da77e190790a45ada9f413c8f8c16542 : SUCCESS in 8m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/feff9f9c51f644b7bc9ad5d839091a66 : SUCCESS in 10m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c49171f57124d6089cfb39c5dee6002 : SUCCESS in 12m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/20014c3b49ce482a9b10483c13712875 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2532e527b03b47de883c469b0e729d32 : SUCCESS in 1h 00m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ceb794e8dfc49ada1a32a631e7a5767 : SUCCESS in 1h 16m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5371ab6c98c4c94a99d4743fbd01b7c : SUCCESS in 1h 27m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/99835546b08741e79637507de7d1788e : SUCCESS in 1h 11m 00s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"0fc391c4601ce9680c8ab680066e0de85722fbcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 18:38:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e86d1ce412f07d04567b966f017b00892377756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 03:37:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40df3a6ca62f4e9788d7731eba07b938 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9032a439daaa4e49bb77235a3e37a9b7 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87ce6ac696664026a62f5576fe1b823c : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/029edf551e19400a80f08d4beaf8e65a : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73466780457c4467aa3d211b80a0dde5 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0c67b2d94b34334ae6173deec9dda73 : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59a1ce3d672649d0987b7aa8de7c4777 : SUCCESS in 7m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd0fd955e01e4d9e9b812c4fb8adb3f9 : SUCCESS in 11m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4bcb262ec5e748ac81e77e54ccb58e5f : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4b20a17c62b04079bc0036e62a2ff41f : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/015d976aad3d4246a79cb4b52b2ec6a1 : SUCCESS in 9m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/811c9ce21a1c49e0b59708e8f5282b03 : SUCCESS in 14m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e066df8004ec41d2b9822d8a6cce37e5 : SUCCESS in 30m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/63f3197d3bc84b1590d52506aaee35ae : SUCCESS in 14m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4b8639b761e84b4485914505abd04345 : SUCCESS in 18m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/258b964d0b24471aaecf39d5d44fd8be : SUCCESS in 18m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3b5560ba79684fd9ad5ea8f420a7d37b : SUCCESS in 5m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/24556284287f495a953c62393ea0c7ae : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b425d47cf8ad4e0eafd9318e1cf4bde6 : SUCCESS in 8m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/39ca51202cbc46789bcab496e43dcebd : SUCCESS in 11m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c0114a9e624e414aba8468c90822f9da : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c858895cdd2a42f595d90796df85d109 : SUCCESS in 58m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edd32652d8e841c7aa0962e7655d2deb : SUCCESS in 1h 15m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ba7584bd6224f24902c00da82e98e85 : TIMED_OUT in 2h 05m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/27b971d63ca04e618f4d11786f4cdce8 : SUCCESS in 1h 13m 57s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"dd72ad8a52a98fc7d079581b0ef8b29d49dd9ae3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-08 09:16:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c49826b06a60ad17dd1824c3a36c971bc91b53fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 12:06:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb7589fc29c47e2bbddc0ef5a7be6f6 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eef24028114d4be3b0037b12f3b2c8f5 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ff896183f3c4442954ebdc729c8aa7b : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf174ff0321b41f38486588d1fe2d7a3 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/115da3af643f4ea8ac9da75d2a7b6674 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80bfd9028c3c46b39326dd101d47be5f : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77b98c1939894b43b9a3af7dd03aec42 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/00209a3608b34811a718375fb9f04b25 : SUCCESS in 12m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ff8818635b2d464dad6f8cfb1fe957e5 : SUCCESS in 16m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6b86ae164190407e8fb3c8e9161643e0 : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/17619688036645fe8ed0a03e66650f43 : SUCCESS in 13m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/32b9a7e9aad14996ba988eb9c32ec9f8 : SUCCESS in 7m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/40e4b4b685b4449491ee98409a02e9d8 : SUCCESS in 12m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/70c58e7d193d49608f944eab78bafcb8 : SUCCESS in 25m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/463e26eef4f948e1b6f26cce4bf554ad : SUCCESS in 19m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/53a457b333e247d398826f97a53b7c0d : SUCCESS in 15m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65c8cd588b424da2a36a01119a9cf861 : SUCCESS in 17m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0af951e6acd747688c095918d71f5210 : SUCCESS in 6m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f26510bbb9f745a88dc2fc22b49476ff : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59d4502ab0b947cbb5bbd9418203f867 : SUCCESS in 8m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b23e0e1f06124c7ba57cb856189d14d8 : SUCCESS in 11m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1311459b59fe444c9c484b6175d53701 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9d6936a49de40a9b581e1d8549e7024 : SUCCESS in 53m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19d7e28fe2b142f5890b68220d7264b0 : SUCCESS in 1h 26m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c0e190672514833a0aedd86c34b8ae8 : SUCCESS in 1h 17m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/799771ec30684a108ce8e22aa08e4dc1 : SUCCESS in 1h 13m 16s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"1db09b265203649aca70689b41112a1d7d9bc260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 12:06:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7025746839083de958e0ecae35fd4ee7bee89bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:32:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00660caabc43418cb11548e1d3f12b8e : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a966bcb77c4225a083c8aa188430e6 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf0cf07733c74e79b9b37b4324516243 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ea277bd1f4f42d695afd47c7e1fb897 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7681e2071004400393f431da43804910 : SUCCESS in 6m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/997be0550d29404c985e70e5b33504bd : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f07f2a1e751a41699c42283c482f0e02 : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f72b762c452e46ecafddb8d7b7a43c10 : SUCCESS in 12m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d744c5fe50ef4e4b83b7b6607d6e5c9a : SUCCESS in 7m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7188031beb534148b2b28f6eb93ad382 : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9067fb4407e14177856765f0e5de625d : SUCCESS in 8m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0cb9ffcc63234559aa949230ac1a48db : SUCCESS in 17m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/954611262a484ad2ba98f1d6994839af : SUCCESS in 21m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/04cb6b87760d405883cf3754d546b69b : SUCCESS in 15m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/957746d2c8284e3aab0e491a512bdfd3 : SUCCESS in 15m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be0af9bae5964ed683d61a5bdbb395a3 : SUCCESS in 18m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/46646a9874e24a6894de023c3e48d062 : SUCCESS in 6m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/03f6263b06ec4a3187a4af437acb256e : SUCCESS in 7m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ed058f98e2db432c95a7bb9de4bd1f56 : SUCCESS in 9m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fd8d9d0bda094018a4fc0204ed642965 : SUCCESS in 14m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0bf512da0334466aac3544608b564b59 : SUCCESS in 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfb26f847c0343bd9501700f126b095f : SUCCESS in 1h 00m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/07f2cb1e129b41c6bbf958e6b23b08ba : SUCCESS in 1h 17m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0025318484147dcbb569f4af32c5cf1 : SUCCESS in 1h 26m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dcdf60be7c6e4e57aff95898ea9fb991 : SUCCESS in 1h 13m 48s\n\nWarning:\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"4fa4d855dabde147157544491889b67e377f2b44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:32:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5ac800ffb4c875bc9e10dc247ed8381ca38511aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:53:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c8b8e9156a5480c97dc3fd9b659ce49 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3155565b2fb79d026c6330559b4f7e73f29986a0","revisions":{"3155565b2fb79d026c6330559b4f7e73f29986a0":{"kind":"REWORK","_number":1,"created":"2020-04-07 10:20:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/718038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/718038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/718038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/718038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/718038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/718038/1"}}},"commit":{"parents":[{"commit":"c75ab1f622598e15d1b51a8d0801384d202630a2","subject":"Merge \"Lower Sphinx constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c75ab1f622598e15d1b51a8d0801384d202630a2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-07 10:20:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-07 10:20:16.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 3.1.1","message":"update constraint for oslo.serialization to new release 3.1.1\n\nChange-Id: I909abcd5a8b8cd52c1df1953533ecacb6e493e95\nmeta:version: 3.1.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ica9d5824b33e3f244d4bff64d6f5d101f5b371fa\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3155565b2fb79d026c6330559b4f7e73f29986a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3155565b2fb79d026c6330559b4f7e73f29986a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
