)]}'
{"id":"openstack%2Frequirements~718043","triplet_id":"openstack%2Frequirements~master~I630e36d530e24e7dc66a4ada3f5eccb724ad8ce6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I630e36d530e24e7dc66a4ada3f5eccb724ad8ce6","subject":"update constraint for oslo.db to new release 8.1.0","status":"MERGED","created":"2020-04-07 10:21:13.000000000","updated":"2020-04-08 19:40:52.000000000","submitted":"2020-04-08 19:32:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718043-1586374366535-49baf0dc","meta_rev_id":"2c770cf9e0fc8bc918301f73bde4621cdf77e77a","_number":718043,"virtual_id_number":718043,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-04-08 09:14:58.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 19:32:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-07 11:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-07 13:14:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-07 13:14:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 11:04:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-07 13:14:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-08 09:14:58.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-08 19:32:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"735888d56e8403c0cb5ef930e1dc732eecb8ee2c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-07 10:21:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15f6d58ac59f76003352955f1d9d7e5f94a7e757","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 11:04:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d4898d5043e8019b11cba4f5ade4aeccc8ced13","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-07 13:14:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da3da9978e839175ecf50c9e159aa87332819205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 18:27:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5b9d559f504706a406ab30c762de2e : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98aed2519c95400b95f5ae57a718a883 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52e9e519eb7b4525b916888af8233aab : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caa978fbf76741f89d809fa7f0470b86 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f549b56511c4734856b9caa13103242 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd75e30f5e71487091e510ca25668c74 : SUCCESS in 10m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b247b0bb58e943e0b94563b76e6f089a : SUCCESS in 6m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3c9c7629a99d4a66826bfcf2aebd25da : SUCCESS in 10m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/67d1c92fb93a4f8daa266007275e29fa : SUCCESS in 13m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2410fefe528e44f8bfce85ff8fdcabf7 : SUCCESS in 8m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c1a080b9d0b7473c9714cbb5a3027cb4 : SUCCESS in 13m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/751a6ce35f974d40a70c573e9e249648 : SUCCESS in 14m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4d666d1211cb49948b1bc7db2e254f3c : SUCCESS in 19m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf02c754e4a54f48834afd7f1fef4e09 : SUCCESS in 24m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7232de006456406ca29d3d775bdaaae2 : SUCCESS in 15m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6f42c095970249e79e1f660701edd972 : SUCCESS in 16m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16a97268fd7a4ff087703dcaeb187cee : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b69b15d031604658b4d8c2b973b81010 : SUCCESS in 7m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2afdb7eb7e5c43fba7a9b6a31373b992 : SUCCESS in 9m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f808ecd827a4d5a8ca78d0b6156da6e : SUCCESS in 10m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d6981fd46dcb4acf930f09791fced005 : SUCCESS in 12m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d1785cc6fdb648ea9669026b4daafa07 : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac347470d9cf4d879f59786ebf0ddfec : SUCCESS in 52m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/180ff201a5124ca5b4f1786391eff4b5 : SUCCESS in 1h 25m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1a4b5bc08174246b2667dca83aaa3da : SUCCESS in 1h 21m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5297232dad114774b696b2fe0b36cd89 : SUCCESS in 1h 07m 52s\n\nWarning:\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"bb8c678f89e78663c21fa3d1d53a2647892765c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 19:50:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40dd10ebe9080fe4bbd98b4dc3df671526674dea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 06:32:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102f2fa9fc3c4fe1a41a25efb4e42b56 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad78180dac44aba9072fe4a94d8c65d : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67f85b8d0e1343c193ee642784a19c08 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c638964bd3554067980a53dda54f18cd : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/915c9dc8c53240109d55cd9697738127 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0dd9ba28aa54826824f11bd159a268e : SUCCESS in 7m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/395c954eb0ca4720be586bd83d743555 : SUCCESS in 9m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2b5f4da3909940c7873ae1da39425700 : SUCCESS in 9m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6d1f1bc3f94e4e58b3f9e72ca8617b24 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9936ed7a6b104ecd8dd0a0b2fdef237f : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86701b8f05484582bed1f3a8c55f3e31 : SUCCESS in 10m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/62260618c3134ccfb81dad5f5b5fd413 : SUCCESS in 14m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a9e465135f4434babc7bc0b673b97bc : SUCCESS in 21m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9638daf09bda46d4bf9e57f7a72b42f4 : SUCCESS in 14m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f47f985eb3bb4acf88e9c9ad9bfd1e21 : SUCCESS in 14m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a69e683b694d43268390cf6ad8c98ee4 : SUCCESS in 18m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8f02f0ed78ae476f96d889842fcac8bb : SUCCESS in 5m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d25dd492cba646318a31df8e3e9014f5 : SUCCESS in 7m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab6970edd79b4db5a25e73ef5495bf39 : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/39703c00d0ef4910a391d1442e21ab05 : SUCCESS in 20m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c762023f7b1e4c2287199020b69bd64d : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d3b6a2b9a83435f86fbec52832faf67 : SUCCESS in 1h 11m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e331fa6e957e44df982a16ed938f9a8e : SUCCESS in 1h 15m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f897ac8f7d824cb388a4ab21bebf8152 : FAILURE in 1h 19m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a04dfccf5387403697f7d282b1492ef8 : SUCCESS in 1h 13m 14s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"4ad58ceb71d953cb4da66b4409c78da9d98967d0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-08 09:14:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a872dc01a08156a293f03e4090ed4d772d668cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:47:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5369631f57f047bf84e2272dfc59e6bc : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc38a081682406dbf7a79c315a68c0f : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/565b57cefec4457fa598d33f7e74c5d4 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fd2598dab304492ab74368f7f04f3eb : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34c83e168c5a4148aed9ebb7101140f3 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/130eee17aaf34ec4b40edfcd666c7836 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/570fbbccaa4c4dd0975a5cfc6dff56dd : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/35b6d2a4febc4368bde925fc8912f326 : SUCCESS in 10m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3dff0b2eedd94d7ab699eb306975cb66 : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a70064fb0d924c1cbc5e591e96b8d7ff : SUCCESS in 6m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5da75880e0a5462688db816c74a7607b : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94364db1e3da4855ab84c540aa4be709 : SUCCESS in 9m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/39bb9a8ed80b476294d2682b5c61f932 : SUCCESS in 14m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/376bf67692ad44baa000fbc764458dc0 : SUCCESS in 29m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6f24e5fc9205444aa17977d827331519 : SUCCESS in 11m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/76a8e9a3a5b240e1982d799341b5d653 : SUCCESS in 15m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c127882df9894f1da6ff04ed1575f5b2 : SUCCESS in 20m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a994c83c189f473fb0c1e0e9d812b185 : SUCCESS in 5m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/046ec325f470420590439ce391ce7834 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7804b2396cc94acd8371bca48668cc99 : SUCCESS in 7m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9fb9d98869294d89ab2528dee3879b27 : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0734a557b16e4f3881fe3aa0f4bc9ce9 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed429bad2b244d93b68fc3ea5df66493 : SUCCESS in 58m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2948594bdaea4a07bbfd6ccdd5585281 : SUCCESS in 1h 22m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d1193c42c6540708d581e24e23c9f88 : SUCCESS in 1h 18m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/517711dfadd94165a48059330e9f6e25 : SUCCESS in 1h 08m 00s\n\nWarning:\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"4ed33b52f382cecc4a66165b11a8398022f38434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:56:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edb450d9128cc49db898db1e771a89a7d8dd02ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:32:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c8aae5bb570e6872e8448b60d21fc293dd75c5de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:32:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3d2ecf562404ce2b0c898fd9c5291a7 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/088bdc3c055647afafa91852ab755bc4 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9463fb9592340ee804afd22f3550ae0 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fa0f827c8094d5fb63af858d0a7a506 : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dc4091e2d1a45d783d59b993b4d2359 : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/955365028d584198b95c7f989285f131 : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cd0286c923554a4fb25c9d889daa9459 : SUCCESS in 12m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1c9662563b36409d9d1662dee57b52d9 : SUCCESS in 10m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c692d499311b4020a2aa2c7f5a028d8c : SUCCESS in 8m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b28c7a5a678945409b1a5aa8daaf8711 : SUCCESS in 13m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/934b3d211f514bb18002406790226663 : SUCCESS in 10m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8eef22d36c7d4f05b40292c142290269 : SUCCESS in 17m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e9f7ac82c5094dcca92e1747a4ddb0c2 : SUCCESS in 33m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf7a0f16d80243299fc4699b05ccf732 : SUCCESS in 28m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e74512dea44f4b829e856d203bb16927 : SUCCESS in 28m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1faa3201e5ec4329b654e3ba2ff3b58b : SUCCESS in 29m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e08314487dcb40069e80e3e900fb40b2 : SUCCESS in 8m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f72dcd0531484a0dbc5044cd3038c87d : SUCCESS in 9m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02a34e5060864797b49a70ccb2e6f8ca : SUCCESS in 10m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f5f1d55c90864ddb84f59e1d3672f71c : SUCCESS in 12m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/193432a7cfdf4996ae72a2c11d4dc5ed : SUCCESS in 15m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18ecbf40162b410281911900c2eaae71 : SUCCESS in 1h 05m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80ee2d8f87954e13bcecb64a314c6323 : SUCCESS in 1h 22m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07e701b10a8d4b6b99d91f829ed25f95 : SUCCESS in 1h 32m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/af66b5de8de741058b794760ca9cac32 : SUCCESS in 1h 13m 40s\n\nWarning:\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/amphorae/backends/utils/keepalivedlvs_query.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_base.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/common/context.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_driver_lib.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-octavia-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path octavia/tests/unit/api/drivers/test_provider_noop_driver.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/test_seagate.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/db/sqlalchemy/api.py in the commit\u0027\n  Job cross-cinder-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py in the commit\u0027\n  Job cross-keystone-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path keystone/tests/unit/test_v3_federation.py in the commit\u0027\n  Job cross-neutron-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path neutron/tests/unit/test_wsgi.py in the commit\u0027\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file neutron/tests/unit/test_wsgi.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file keystone/tests/unit/test_v3_federation.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_seagate.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file cinder/db/sqlalchemy/api.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/dell_emc/unity/test_replication.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":1},{"id":"694a55bc6c888b2e602f28cb018f67cdf2956331","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:40:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/429898fe38cc4bd1bd108abf74891b62 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c2a6dcdbad45420f0e1035fb538b66909015b1d","revisions":{"3c2a6dcdbad45420f0e1035fb538b66909015b1d":{"kind":"REWORK","_number":1,"created":"2020-04-07 10:21:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/718043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/718043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/718043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/718043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/718043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/718043/1"}}},"commit":{"parents":[{"commit":"c75ab1f622598e15d1b51a8d0801384d202630a2","subject":"Merge \"Lower Sphinx constraint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c75ab1f622598e15d1b51a8d0801384d202630a2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-07 10:21:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-07 10:21:11.000000000","tz":0},"subject":"update constraint for oslo.db to new release 8.1.0","message":"update constraint for oslo.db to new release 8.1.0\n\nChange-Id: I630e36d530e24e7dc66a4ada3f5eccb724ad8ce6\nmeta:version: 8.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ica9d5824b33e3f244d4bff64d6f5d101f5b371fa\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c2a6dcdbad45420f0e1035fb538b66909015b1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c2a6dcdbad45420f0e1035fb538b66909015b1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
