)]}'
{"id":"openstack%2Frequirements~718202","triplet_id":"openstack%2Frequirements~stable%2Focata~I7f372e977d3c61f1a680866e10b2b7a429a99ae3","project":"openstack/requirements","branch":"stable/ocata","topic":"718202","hashtags":[],"change_id":"I7f372e977d3c61f1a680866e10b2b7a429a99ae3","subject":"[ocata only] Add stestr to upper constraints","status":"ABANDONED","created":"2020-04-07 20:01:18.000000000","updated":"2020-05-28 17:00:18.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f89bfc24514368d41b4c318a608cc2f3fba47f8","_number":718202,"virtual_id_number":718202,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-27 23:19:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 02:54:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-11 22:36:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-12 16:23:57.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2020-05-27 19:53:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-05-27 23:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f23ffd72a018c7e609ccc477e65e31bae142bfd2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-07 20:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5531eeae872fd19ebf60cd8a8c00c5b75b1286d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:56:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9ad9f9929bc04a989f16708ed59ee4be : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e7409428bc14a0687917da4d6e2a813 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cc03e17b8744abd91fa6f99c73ee5e3 : SUCCESS in 5m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dae2165ec6894f0f8801b232ed62b9c8 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7fd19bee6ce4600940cf32552bb84da : SUCCESS in 7m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4384f2e41754580a249bf65f9810d26 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5d8865b312444b069c52caf078017f60 : FAILURE in 10m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b403383684334436b32427934e0a5185 : SUCCESS in 14m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9c71f03b9fa44d42a488d91c70387184 : SUCCESS in 15m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6bfe90b0cddb422e9f5a6e7fc64db8b6 : SUCCESS in 10m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2edc577a8b484a41b19e113916c49c35 : SUCCESS in 13m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7d5c467f7fea4fb3a02477666df00849 : SUCCESS in 22m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ad38ae24600c4c0780cb741c3fa2a7b0 : SUCCESS in 16m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f401c49ef104a09b3059cabe255f624 : SUCCESS in 20m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/20f6b93cea8c410eb26e4901f50e3f14 : SUCCESS in 9m 13s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/18392c59151140a6b0db2ae1a3ec890d : FAILURE in 36m 02s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e7891020b636412699706cbc6a522f68 : SUCCESS in 1h 25m 44s\n\nWarning:\n  Job cross-cinder-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/volume/drivers/ibm/ibm_storage/proxy.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/common/wsgi.py in the commit\u0027\n  Comments left for invalid file glance/common/wsgi.py\n  Comments left for invalid file cinder/volume/drivers/ibm/ibm_storage/proxy.py","accounts_in_message":[],"_revision_number":1},{"id":"f40c3b3163088154111147171ae938b1c1b2d6e2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-08 02:54:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0c8b5eaca6270f1d1dfc36dd953d4b8f8450b7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 04:31:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5f51ef41eae34f35944eee484166d03a : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7962bbb784dd43d2b211ad113cbbfc46 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c5f0915adc242fb8b7dd318cf35c820 : SUCCESS in 9m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eebfe252dde54210bcede6e32ac5bf2f : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6755dce9c02a4e309d694da858bfeac6 : SUCCESS in 7m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4017ffd467404c9bad8285ab99d93515 : SUCCESS in 6m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/df5f0a1fed4f48e6b8ab7dc31ad95fab : FAILURE in 10m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/18bed176e3a94563aaee849e6bcaff22 : SUCCESS in 9m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b3b932c9f4824a239570d66b85507c2b : SUCCESS in 10m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2aa147ad083a47a8aa8e8422015147cf : SUCCESS in 10m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4f0944b3ad52471d8978d05b1fa9b234 : SUCCESS in 15m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8cec5c8de1864284939eded219d2c28c : SUCCESS in 22m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/93027346018c403f9dc99fb126784b87 : SUCCESS in 11m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d905e4295481451ba28a88560e9f779f : SUCCESS in 16m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/578925744bd14c428057e02179fd003b : SUCCESS in 9m 49s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e020dd08e30f4c17a2565506ee43b54b : FAILURE in 30m 08s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/cdeb136e3dd64653b5cbef727bc59717 : SUCCESS in 1h 33m 00s\n\nWarning:\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/common/wsgi.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-cinder-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/volume/drivers/ibm/ibm_storage/proxy.py in the commit\u0027\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/common/wsgi.py\n  Comments left for invalid file cinder/volume/drivers/ibm/ibm_storage/proxy.py","accounts_in_message":[],"_revision_number":1},{"id":"20b8d103522a5145c1e4b8ffaa5683ecb4416828","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-08 06:34:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96f4a790cd0bcd3dfa1d17fcc484f55f95927664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:56:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/11f4f481c0af472fb9a40a3e36a520e0 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003f4ec1c83b422ba9fcaee9411ca244 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3df144b9eea740159a5f1bbf856ecab5 : SUCCESS in 6m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a200b0280c24e01a917a736eff2fad3 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a97bbae4403946b0aecb8537e84ab757 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/260fbe0b48674815be6c843740d581ef : SUCCESS in 7m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/883da91424ae4bb8b1b55adf94c91a86 : FAILURE in 11m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/186f84b5dd174b3db4cdb4bf76d832a0 : SUCCESS in 10m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/35ccab5bae2c4e69bbf26ba02096f8e5 : SUCCESS in 14m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e327371f368f45c98b83398bc7361812 : SUCCESS in 12m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/57a6ae5d96e045848a325c81a82fe4bd : SUCCESS in 13m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e5210e9f1b5643729e6c7e8816d8b9f8 : SUCCESS in 23m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/00a1216973534147b8b4d28dfdc8e001 : SUCCESS in 13m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4e5d8839ecf40d3b8fa580174efb5b8 : SUCCESS in 22m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6dcec11b02104e5ca4481d9cd5063db1 : SUCCESS in 9m 30s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9e265ce8fed34181acd89ecdc3e9b7f0 : FAILURE in 34m 02s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7ef40ea3a5684dd6a02ab78374d524a0 : FAILURE in 1h 13m 24s\n\nWarning:\n  Job cross-cinder-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path cinder/volume/drivers/ibm/ibm_storage/proxy.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/common/wsgi.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/common/test_timeutils.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Job cross-glance-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path glance/tests/unit/v2/test_tasks_resource.py in the commit\u0027\n  Comments left for invalid file glance/common/wsgi.py\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/tests/unit/common/test_timeutils.py\n  Comments left for invalid file cinder/volume/drivers/ibm/ibm_storage/proxy.py","accounts_in_message":[],"_revision_number":2},{"id":"b4eb6654527e8b8ebd91901a72c99c7ef5f0c315","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-08 14:06:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1762f87c50c9a2e89725544b2702bf6dae48b9e2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-12 17:03:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ff63cffe0890627f55980947ff765db035d201e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 17:40:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/66cdbed3ec6a48a89fc069d221da9e8f : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2243382682e446080f1e885a14a5447 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b737bf3c4bc14f9a84011bcdf5989392 : SUCCESS in 5m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3748c58dfc31470c94a5941263ee5099 : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0532376271554d3fb517a73cb4018ee4 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/983c02a6a397479990aee6defe514fd0 : SUCCESS in 4m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e9ba31f8e0f84787b2acb6324ab9ba5b : SUCCESS in 9m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/477665abc9444f47a03105e071df7be3 : SUCCESS in 8m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c5f3dbe278a24c6eaa4bf4033cfa7a81 : FAILURE in 11m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a66401a3ec6a4ab4887812241703c514 : SUCCESS in 8m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0f82cc9ea6854b9cab94d5f6b312015e : SUCCESS in 27m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/59a06cd4a68648fb9ea18950630805ba : SUCCESS in 19m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9a42ecb4c2894aaf9a6c67b912bcf7d6 : SUCCESS in 14m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8927eb05721440db522f0e85340f344 : SUCCESS in 15m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/214081dc2c714172b8a5b059932732a8 : SUCCESS in 9m 37s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/20f50d492c504f6a89457a2ee5b7c98b : FAILURE in 30m 52s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/460f1237097f4d37bf30afccc58979c6 : FAILURE in 24m 55s","accounts_in_message":[],"_revision_number":3},{"id":"99cb472390f6a0ced534353a1b9fff98c0e3987e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-14 12:17:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9cb91175a7447f6d4c22e7bf447a86781d05f457","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-14 12:29:45.000000000","message":"Patch Set 4:\n\nI thought projects were still using ostestr in ocata?","accounts_in_message":[],"_revision_number":4},{"id":"fecced405d35ad5327cc6df6ff51f48bc07593af","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-14 12:45:24.000000000","message":"Patch Set 4:\n\n\u003e I thought projects were still using ostestr in ocata?\n\nI remember something like that, too, but in fact it seems (at least in dsvm tempest jobs) there is already stestr installed. Moreover, stestr\u003d\u003d3.0.0 (see [1]) :(\n\nAnd the same problem is in grenade jobs for Pike [2], which installs Ocata first and tries to run smoke test from tempest.\n\n[1] https://review.opendev.org/#/c/714958/\n[2] https://review.opendev.org/#/c/669538/","accounts_in_message":[],"_revision_number":4},{"id":"4e340a131a3704030eb356311b635b2a2e3f8f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 20:39:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4e276ab0529f4fa3a32f221d0ca6ae10 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cffd637114aa45508b5afea0592e6c6f : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49aeb3f9ba014ff595b678a4b1ebbd1c : SUCCESS in 7m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/502213fc1eb24c778878c963859d4d26 : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70dcc392f1734cbea1dd88185ffa02a7 : SUCCESS in 6m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5be1451fce648cb83a3309aeb4daa81 : SUCCESS in 6m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fa1a29c5609c4c35ac4aae5e92c42507 : SUCCESS in 9m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/56f4136a93474b69a06bb2e20845bd88 : SUCCESS in 10m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a7f174d4d9d7428fa65cecb92590e132 : SUCCESS in 14m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/331560e1746b418e949dd124563d6a09 : SUCCESS in 10m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c139661556384c7a92231d14cb70f509 : TIMED_OUT in 41m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/75f36a4566af42f9a7183c0bb44e9b08 : SUCCESS in 20m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6bdf9a557ee54dca9f0fa6da0d578901 : SUCCESS in 15m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96ed19d321fe40b69015233dabd5b851 : SUCCESS in 17m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/014c64655e36450291c8f75ed0ef8402 : SUCCESS in 11m 24s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/2d86136753954a9ab9f343a04732b942 : FAILURE in 44m 06s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/c08b0b0d5a454310813bb92c6916821a : FAILURE in 41m 30s","accounts_in_message":[],"_revision_number":4},{"id":"188a2e7fba79199a861c5923f8388d7d7aa92139","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 21:47:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"212a9d7db3b2c6316a0d51e9d71dc253350ef418","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 21:48:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fe8f93fe0010bb757c3698ecfadf8e7b5028ac6c","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 22:09:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e38fb9703b310a23c68bd91df2bc38efcd63757b","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 22:12:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7c85e5446af442dac593d63148bd0365bd58c167","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 22:36:03.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9aae9956431f02ff3fc3b142bafaee26a088bb3f","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 22:56:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"037969104bc9e86cc139e083802f9e7340a5e6c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:57:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f15660b3f183410c91af473d721ecbf7 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c3bff6b1dc4f108541ded8efe77871 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55ead3e49e464d1fa4ff93371190bc03 : SUCCESS in 8m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c498ba2a083e4716a91914358cf835ba : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b75e27bdbf24e35ab6463c809557291 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e664a847c89a4fe4be4040dfc53fe3e5 : SUCCESS in 6m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/17a16fab14244510a013448837845815 : SUCCESS in 9m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6a1210bfd1d140e295d2f5b8d4121990 : FAILURE in 12m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d3af4015cb1d424990878e5f333f73ce : SUCCESS in 14m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f4ba2a5587e74ad189a9b33e95b30add : SUCCESS in 12m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d98b072d257d47ef83534af38a983fe0 : SUCCESS in 27m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/acc0db646ed4469f8b9dfbfc94dc5a6a : SUCCESS in 28m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6f076832cadc434ba5509760507ee04f : SUCCESS in 24m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb0a3344d0b54a7fb433b930e1de219d : SUCCESS in 25m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/772e58cedffc45918eb22091cbf99f34 : SUCCESS in 10m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/95a25c4fbe13486fba063da38dcf563b : FAILURE in 41m 21s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/eb26922341ce461a91f521fefeacf923 : FAILURE in 29m 58s","accounts_in_message":[],"_revision_number":5},{"id":"4575b6f8b731a9d1420e2bbcba56c8b78407b298","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 23:48:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e7303f10bc99d47d0c4af5e5f1ea6debc02b87d5","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 23:51:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"02dc1e323a723ea19b68abc98798dee41fda1972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 01:23:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a1d22a8a844f41e3a9706c0a1c92b14b : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3926d3c6c74414b9ba44406ca7cd50a : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d24c9244b9704f2891e11fb799c2a8dd : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1abfc0d93d944e008186b3df770fa619 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a1bacdb4618437cad669deb309253a6 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ad6e3d358f74ed5bd40808f71c1585e : SUCCESS in 3m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b13d9a001c694c0fa5ff51c2b07d0103 : FAILURE in 7m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/578d40ea7e7f4208b1bbd39d0832e17d : SUCCESS in 8m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e4dc5d4652f849678f6651e7c4f9d367 : SUCCESS in 14m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/abc71662c251499cace17fb02aeb2e60 : SUCCESS in 10m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/13fd28ce1e91492db3fcbdfb40c7829f : SUCCESS in 12m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2e14feff58db4e3caba7c139e4763868 : SUCCESS in 25m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3a799e4b362b46b2a6ee04a906aa0d11 : SUCCESS in 12m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf4cb9fbf83e46b4982d3df857f771f6 : SUCCESS in 16m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/038220d17c9041ecb8788f6cc721a69e : SUCCESS in 7m 50s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7334c59d02944b149d062ea7b1f46e8d : FAILURE in 38m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0e4cb8bb82594b9096778fe8ed608025 : FAILURE in 1h 31m 21s","accounts_in_message":[],"_revision_number":6},{"id":"36ad96c2ec3e8da5db39c54683801236fd450d65","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-12 03:30:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1cf8f330ac4bcf5645179efb8331277681c3229f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 05:06:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/73b254b9ec30419685d383a476ccc606 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/685a1f3303864bdea2975f2a6a57d71c : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11fa8467d953491ea4b450c0a15a0ba9 : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a46bd47b42bc431e8ce531c4e8d36042 : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08c87fa3f4994b9abd168cb6e5755087 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa06525e28684d219a36e49d0aac8c1e : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d21bb83f56a1499a9327129ba7f88176 : FAILURE in 9m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6de05bc5275a477aa302e1bf578cb8f3 : SUCCESS in 8m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9830047de53f44bebfc5d5ba75b91ed2 : SUCCESS in 10m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/be13ee95490245dbafc91227bf1631a2 : SUCCESS in 10m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/87498a57d3b2483589c03283c46588ea : SUCCESS in 12m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6b8e32c1b89f4ed9a15e11896adb4907 : SUCCESS in 20m 13s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7a18541491c94b07a1c300b7aced9969 : SUCCESS in 14m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b3a880b321f4e0eba8ceb289b39cea6 : SUCCESS in 17m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/350fd9e878f6447ba1f5e60d95657a7b : SUCCESS in 8m 09s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/09d3037f67464b658b391b076156d901 : FAILURE in 35m 44s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/062696477f694e95b4a7124a3c4d8ff9 : SUCCESS in 1h 30m 45s","accounts_in_message":[],"_revision_number":7},{"id":"06819887285255da671c0e1f408a3abab83c2839","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-12 05:43:15.000000000","message":"Patch Set 7:\n\nSo looks like we are royally screwed at the moment. There\u0027s no way we can get past the \u0027legacy-tempest-dsvm-neutron-full\u0027 gate without stestr minimal version 1.0.0 as that\u0027s the version where the stestr module starting to export the \u0027run_command\u0027 which tempest dependes on. However, with stestr version 1.0.0, it requires pbr version 2.0.0 or greater. This will fail the \u0027requirements-tox-py27-check-uc\u0027 gate as the upper constraint for pbr is 1.0.0. Besides, the \u0027requirements-integration\u0027 gate is also failing with weird directory permission issue.","accounts_in_message":[],"_revision_number":7},{"id":"d89b16354bf16174221f365b8ee00a06aacad6a2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-12 08:29:34.000000000","message":"Patch Set 7:\n\n\u003e So looks like we are royally screwed at the moment. There\u0027s no way\n \u003e we can get past the \u0027legacy-tempest-dsvm-neutron-full\u0027 gate without\n \u003e stestr minimal version 1.0.0 as that\u0027s the version where the stestr\n \u003e module starting to export the \u0027run_command\u0027 which tempest dependes\n \u003e on. However, with stestr version 1.0.0, it requires pbr version\n \u003e 2.0.0 or greater. This will fail the \u0027requirements-tox-py27-check-uc\u0027\n \u003e gate as the upper constraint for pbr is 1.0.0. Besides, the\n \u003e \u0027requirements-integration\u0027 gate is also failing with weird\n \u003e directory permission issue.\n\nThanks Guang Yee for looking into this! One thing maybe worth to check is which version of tempest is used in the tempest-dsvm-neutron-full job (actually, devstack or devstack-gate) as it seems maybe the same version (20.0.0?) that meant to be for Stein. We might need to check with Gmann about this. (Unfortunately I don\u0027t have time right now to look into this, so I really appreciate you\u0027ve looked at it so far)","accounts_in_message":[],"_revision_number":7},{"id":"232c1ea3b7535a3a9505ea87eb5415d7e8be4173","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-12 16:23:57.000000000","message":"Patch Set 7:\n\n\u003e \u003e So looks like we are royally screwed at the moment. There\u0027s no\n \u003e way\n \u003e \u003e we can get past the \u0027legacy-tempest-dsvm-neutron-full\u0027 gate\n \u003e without\n \u003e \u003e stestr minimal version 1.0.0 as that\u0027s the version where the\n \u003e stestr\n \u003e \u003e module starting to export the \u0027run_command\u0027 which tempest\n \u003e dependes\n \u003e \u003e on. However, with stestr version 1.0.0, it requires pbr version\n \u003e \u003e 2.0.0 or greater. This will fail the \u0027requirements-tox-py27-check-uc\u0027\n \u003e \u003e gate as the upper constraint for pbr is 1.0.0. Besides, the\n \u003e \u003e \u0027requirements-integration\u0027 gate is also failing with weird\n \u003e \u003e directory permission issue.\n \u003e \n \u003e Thanks Guang Yee for looking into this! One thing maybe worth to\n \u003e check is which version of tempest is used in the tempest-dsvm-neutron-full\n \u003e job (actually, devstack or devstack-gate) as it seems maybe the\n \u003e same version (20.0.0?) that meant to be for Stein. We might need to\n \u003e check with Gmann about this. (Unfortunately I don\u0027t have time right\n \u003e now to look into this, so I really appreciate you\u0027ve looked at it\n \u003e so far)\n\nFrom the logs looks like we are always cloning/using tempest master branch even though it is pinned in upper-constraints.txt (tempest\u003d\u003d\u003d14.0.0). I honestly don\u0027t know why the gate is setup that way.","accounts_in_message":[],"_revision_number":7},{"id":"9d84370e199306722e808e64ceae67e6a7734738","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-05-27 19:53:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"63ce5058649ae4b56906f5eb11c292da1f1d2bde","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-27 21:11:23.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"a8baf84495ed61f398d4f704bccfef7ed2692ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:19:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/41e4be4714454dfdaf4261ba14b13b52 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f232639445047518d5b6188a61bd398 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b25d6f1e41a4c4b8fc5a6f4a3af93cb : SUCCESS in 5m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aa7a16b2a0d44326b30478e791054b76 : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3efc197a357749beaf06be38e221a554 : SUCCESS in 7m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a72395eec2e4f92a2ffc33e8a900d52 : SUCCESS in 7m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e324e2d9286d4ae385ad360b57069606 : FAILURE in 8m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c46bc3ad6bcd45008c82d71ac0309691 : SUCCESS in 9m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/49554963099f455094159fde15390ce0 : SUCCESS in 13m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/78a6c4fe91c345c498d67b0df9dd642a : SUCCESS in 9m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/140a82fa380a4ff7b77ba562dd2545fb : SUCCESS in 15m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/36b4ebfecd6542f2acc83734b67ab7a5 : SUCCESS in 34m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/afa20b88aed148f4a023008905ce3e8e : SUCCESS in 22m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5485ddb7bd264416867ca4264f728411 : SUCCESS in 19m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/718a6c9c2a3c46899217c1b1a4637f9c : SUCCESS in 8m 25s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/10c6caf4430341babe6a834ee30b8af6 : FAILURE in 1h 23m 50s","accounts_in_message":[],"_revision_number":8},{"id":"df0c4a12f5a49c857c1fea6ee1dfd3fed3b2d90b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 17:00:18.000000000","message":"Abandoned\n\nthis is now done (as part of the patchset that fixed ocata gating)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"2f2d40c7882eabc76b059ca503fc0ede6242b45a","revisions":{"d1343ddba638f7152955660d02461f532c9e73e0":{"kind":"REWORK","_number":1,"created":"2020-04-07 20:01:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/718202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718202/1"}}},"commit":{"parents":[{"commit":"831cbb687a7179d7591f29699bcd8fe09a534aeb","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831cbb687a7179d7591f29699bcd8fe09a534aeb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"subject":"[ocata only] Add stestr to upper constraints","message":"[ocata only] Add stestr to upper constraints\n\nStestr dropped python 2 support with version 3.0.0, but setup.cfg was\nwrongly configured in that release [1], which was fixed in version\n3.0.1. Unfortunately this causes issues in python 2 as pip falsely\ninstalls 3.0.0 as it is not restricted as python 3 only yet.\n\nIn Ocata there isn\u0027t any constraint for stestr, but grenade job on\nPike installs Tempest with using upper-constraints.txt from Ocata,\nwhich does not cap stestr and so pip installs version 3.0.0, and\ngrenade job fails with the error when running stestr:\n\nImportError: No module named configparser\n\nEither stestr 3.0.0 needs to be backlisted or needs to be constrained.\nThis patch backports the stestr\u003d\u003d\u003d2.0.0 upper constraint.\n\n[1] https://github.com/mtreinish/stestr/issues/292\n\nChange-Id: I7f372e977d3c61f1a680866e10b2b7a429a99ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1343ddba638f7152955660d02461f532c9e73e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1343ddba638f7152955660d02461f532c9e73e0"}]},"branch":"refs/heads/stable/ocata"},"499c5cc80d2c6d24cddbafa7a06f7bd3009c01e4":{"kind":"REWORK","_number":2,"created":"2020-04-08 06:34:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/718202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718202/2"}}},"commit":{"parents":[{"commit":"831cbb687a7179d7591f29699bcd8fe09a534aeb","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831cbb687a7179d7591f29699bcd8fe09a534aeb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-08 06:33:41.000000000","tz":120},"subject":"[ocata only] Add stestr to upper constraints","message":"[ocata only] Add stestr to upper constraints\n\nStestr dropped python 2 support with version 3.0.0, but setup.cfg was\nwrongly configured in that release [1], which was fixed in version\n3.0.1. Unfortunately this causes issues in python 2 as pip falsely\ninstalls 3.0.0 as it is not restricted as python 3 only yet.\n\nIn Ocata there isn\u0027t any constraint for stestr, but grenade job on\nPike installs Tempest with using upper-constraints.txt from Ocata,\nwhich does not cap stestr and so pip installs version 3.0.0, and\ngrenade job fails with the error when running stestr:\n\nImportError: No module named configparser\n\nEither stestr 3.0.0 needs to be backlisted or needs to be constrained.\nThis patch sets the stestr\u003d\u003d\u003d1.1.0 upper constraint.\n\n[1] https://github.com/mtreinish/stestr/issues/292\n\nChange-Id: I7f372e977d3c61f1a680866e10b2b7a429a99ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/499c5cc80d2c6d24cddbafa7a06f7bd3009c01e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/499c5cc80d2c6d24cddbafa7a06f7bd3009c01e4"}]},"branch":"refs/heads/stable/ocata"},"3f2762f1744c8fce1ecf64cc3f17720b42c169f0":{"kind":"REWORK","_number":3,"created":"2020-04-12 17:03:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/718202/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718202/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718202/3"}}},"commit":{"parents":[{"commit":"831cbb687a7179d7591f29699bcd8fe09a534aeb","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831cbb687a7179d7591f29699bcd8fe09a534aeb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-08 14:00:11.000000000","tz":120},"subject":"[ocata only] Add stestr to upper constraints","message":"[ocata only] Add stestr to upper constraints\n\nStestr dropped python 2 support with version 3.0.0, but setup.cfg was\nwrongly configured in that release [1], which was fixed in version\n3.0.1. Unfortunately this causes issues in python 2 as pip falsely\ninstalls 3.0.0 as it is not restricted as python 3 only yet.\n\nIn Ocata there isn\u0027t any constraint for stestr, but grenade job on\nPike installs Tempest with using upper-constraints.txt from Ocata,\nwhich does not cap stestr and so pip installs version 3.0.0, and\ngrenade job fails with the error when running stestr:\n\nImportError: No module named configparser\n\nEither stestr 3.0.0 needs to be backlisted or needs to be constrained.\nThis patch sets the stestr\u003d\u003d\u003d0.4.0 upper constraint and adjust other\ndependencies to make the tests pass.\n\n[1] https://github.com/mtreinish/stestr/issues/292\n\nChange-Id: I7f372e977d3c61f1a680866e10b2b7a429a99ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f2762f1744c8fce1ecf64cc3f17720b42c169f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f2762f1744c8fce1ecf64cc3f17720b42c169f0"}]},"branch":"refs/heads/stable/ocata"},"cb0d5b0e381c8aea0c1df5016a278ca2527e5cd0":{"kind":"REWORK","_number":4,"created":"2020-04-14 12:17:47.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/718202/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718202/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718202/4"}}},"commit":{"parents":[{"commit":"831cbb687a7179d7591f29699bcd8fe09a534aeb","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831cbb687a7179d7591f29699bcd8fe09a534aeb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-14 12:17:01.000000000","tz":120},"subject":"[ocata only] Add stestr to upper constraints","message":"[ocata only] Add stestr to upper constraints\n\nStestr dropped python 2 support with version 3.0.0, but setup.cfg was\nwrongly configured in that release [1], which was fixed in version\n3.0.1. Unfortunately this causes issues in python 2 as pip falsely\ninstalls 3.0.0 as it is not restricted as python 3 only yet.\n\nIn Ocata there isn\u0027t any constraint for stestr, but grenade job on\nPike installs Tempest with using upper-constraints.txt from Ocata,\nwhich does not cap stestr and so pip installs version 3.0.0, and\ngrenade job fails with the error when running stestr:\n\nImportError: No module named configparser\n\nEither stestr 3.0.0 needs to be backlisted or needs to be constrained.\nThis patch sets the stestr\u003d\u003d\u003d0.4.0 upper constraint and adjust other\ndependencies to make the tests pass.\n\n[1] https://github.com/mtreinish/stestr/issues/292\n\nChange-Id: I7f372e977d3c61f1a680866e10b2b7a429a99ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb0d5b0e381c8aea0c1df5016a278ca2527e5cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb0d5b0e381c8aea0c1df5016a278ca2527e5cd0"}]},"branch":"refs/heads/stable/ocata"},"7e115794c871be88f1d3b6c14e8d30c7c33c2f55":{"kind":"REWORK","_number":5,"created":"2020-05-11 22:12:53.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/02/718202/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718202/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718202/5"}}},"commit":{"parents":[{"commit":"831cbb687a7179d7591f29699bcd8fe09a534aeb","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831cbb687a7179d7591f29699bcd8fe09a534aeb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-05-11 22:12:52.000000000","tz":-420},"subject":"[ocata only] Add stestr to upper constraints","message":"[ocata only] Add stestr to upper constraints\n\nStestr dropped python 2 support with version 3.0.0, but setup.cfg was\nwrongly configured in that release [1], which was fixed in version\n3.0.1. Unfortunately this causes issues in python 2 as pip falsely\ninstalls 3.0.0 as it is not restricted as python 3 only yet.\n\nIn Ocata there isn\u0027t any constraint for stestr, but grenade job on\nPike installs Tempest with using upper-constraints.txt from Ocata,\nwhich does not cap stestr and so pip installs version 3.0.0, and\ngrenade job fails with the error when running stestr:\n\nImportError: No module named configparser\n\nEither stestr 3.0.0 needs to be backlisted or needs to be constrained.\nThis patch sets the stestr\u003d\u003d\u003d0.4.0 upper constraint and adjust other\ndependencies to make the tests pass.\n\n[1] https://github.com/mtreinish/stestr/issues/292\n\nChange-Id: I7f372e977d3c61f1a680866e10b2b7a429a99ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e115794c871be88f1d3b6c14e8d30c7c33c2f55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e115794c871be88f1d3b6c14e8d30c7c33c2f55"}]},"branch":"refs/heads/stable/ocata"},"b3e1a72349991d7ee7e00f87e194bce15811c09c":{"kind":"REWORK","_number":6,"created":"2020-05-11 23:51:04.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/02/718202/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718202/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718202/6"}}},"commit":{"parents":[{"commit":"831cbb687a7179d7591f29699bcd8fe09a534aeb","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831cbb687a7179d7591f29699bcd8fe09a534aeb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-05-11 23:50:56.000000000","tz":-420},"subject":"[ocata only] Add stestr to upper constraints","message":"[ocata only] Add stestr to upper constraints\n\nStestr dropped python 2 support with version 3.0.0, but setup.cfg was\nwrongly configured in that release [1], which was fixed in version\n3.0.1. Unfortunately this causes issues in python 2 as pip falsely\ninstalls 3.0.0 as it is not restricted as python 3 only yet.\n\nIn Ocata there isn\u0027t any constraint for stestr, but grenade job on\nPike installs Tempest with using upper-constraints.txt from Ocata,\nwhich does not cap stestr and so pip installs version 3.0.0, and\ngrenade job fails with the error when running stestr:\n\nImportError: No module named configparser\n\nEither stestr 3.0.0 needs to be backlisted or needs to be constrained.\nThis patch sets the stestr\u003d\u003d\u003d0.4.0 upper constraint and adjust other\ndependencies to make the tests pass.\n\n[1] https://github.com/mtreinish/stestr/issues/292\n\nChange-Id: I7f372e977d3c61f1a680866e10b2b7a429a99ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3e1a72349991d7ee7e00f87e194bce15811c09c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3e1a72349991d7ee7e00f87e194bce15811c09c"}]},"branch":"refs/heads/stable/ocata"},"7153973eeaa5cb5c3090f25537fed103380fe3db":{"kind":"REWORK","_number":7,"created":"2020-05-12 03:30:36.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/02/718202/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718202/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718202/7"}}},"commit":{"parents":[{"commit":"831cbb687a7179d7591f29699bcd8fe09a534aeb","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831cbb687a7179d7591f29699bcd8fe09a534aeb"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-05-12 03:30:24.000000000","tz":-420},"subject":"[ocata only] Add stestr to upper constraints","message":"[ocata only] Add stestr to upper constraints\n\nStestr dropped python 2 support with version 3.0.0, but setup.cfg was\nwrongly configured in that release [1], which was fixed in version\n3.0.1. Unfortunately this causes issues in python 2 as pip falsely\ninstalls 3.0.0 as it is not restricted as python 3 only yet.\n\nIn Ocata there isn\u0027t any constraint for stestr, but grenade job on\nPike installs Tempest with using upper-constraints.txt from Ocata,\nwhich does not cap stestr and so pip installs version 3.0.0, and\ngrenade job fails with the error when running stestr:\n\nImportError: No module named configparser\n\nEither stestr 3.0.0 needs to be backlisted or needs to be constrained.\nThis patch sets the stestr\u003d\u003d\u003d0.4.0 upper constraint and adjust other\ndependencies to make the tests pass.\n\n[1] https://github.com/mtreinish/stestr/issues/292\n\nChange-Id: I7f372e977d3c61f1a680866e10b2b7a429a99ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7153973eeaa5cb5c3090f25537fed103380fe3db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7153973eeaa5cb5c3090f25537fed103380fe3db"}]},"branch":"refs/heads/stable/ocata"},"2f2d40c7882eabc76b059ca503fc0ede6242b45a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-05-27 21:11:23.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/02/718202/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718202/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718202/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718202/8"}}},"commit":{"parents":[{"commit":"ede3aafb0bee2d20b5dde905f4116dd4898908f4","subject":"Fix job issues for stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ede3aafb0bee2d20b5dde905f4116dd4898908f4"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-07 19:36:47.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-27 21:11:23.000000000","tz":0},"subject":"[ocata only] Add stestr to upper constraints","message":"[ocata only] Add stestr to upper constraints\n\nStestr dropped python 2 support with version 3.0.0, but setup.cfg was\nwrongly configured in that release [1], which was fixed in version\n3.0.1. Unfortunately this causes issues in python 2 as pip falsely\ninstalls 3.0.0 as it is not restricted as python 3 only yet.\n\nIn Ocata there isn\u0027t any constraint for stestr, but grenade job on\nPike installs Tempest with using upper-constraints.txt from Ocata,\nwhich does not cap stestr and so pip installs version 3.0.0, and\ngrenade job fails with the error when running stestr:\n\nImportError: No module named configparser\n\nEither stestr 3.0.0 needs to be backlisted or needs to be constrained.\nThis patch sets the stestr\u003d\u003d\u003d0.4.0 upper constraint and adjust other\ndependencies to make the tests pass.\n\n[1] https://github.com/mtreinish/stestr/issues/292\n\nChange-Id: I7f372e977d3c61f1a680866e10b2b7a429a99ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2d40c7882eabc76b059ca503fc0ede6242b45a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2d40c7882eabc76b059ca503fc0ede6242b45a"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
