)]}'
{"id":"openstack%2Frequirements~718664","triplet_id":"openstack%2Frequirements~master~Ia9db5f04b4c6738bd25de11ebc46af85e51aedd6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia9db5f04b4c6738bd25de11ebc46af85e51aedd6","subject":"update constraint for python-qinlingclient to new release 5.0.1","status":"MERGED","created":"2020-04-09 11:08:45.000000000","updated":"2020-04-09 20:30:12.000000000","submitted":"2020-04-09 20:27:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718664-1586464035559-a17bf472","meta_rev_id":"a2db2b0eb87b06f4add7fc5e408a5214d40977af","_number":718664,"virtual_id_number":718664,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 20:27:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-09 18:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-09 18:39:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 18:39:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-09 20:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ca78779b87d307dff39a7889e198c72be3cb4df","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 11:08:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37589eaa99733e607ba98c71e1a7be1dda577868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 16:34:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c9ddb33f1a446c693b29b2de5948856 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf78b26aeaa4adbaf27f2909066966b : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73d9690cfbfe4e6f9b703991ca95c5f6 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8c5f03561c741c28807085c2d60e1eb : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6449fa1a2e0749f09cb1661239720e48 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7f8130bc9834b1586c64d656bcdc96e : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32e981d9206a4a48a573af337a3081b2 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d94366bd274486d9d2c4847c2f0e3e0 : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b1eafc09c5ba487bbc2545bea65cb9be : SUCCESS in 10m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/19c160d568b3488d95d683d37bd2d8e1 : SUCCESS in 5m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db3661f3925f41e7b01187352106e34d : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94ae65e522a14ecd9ee41e6263a9dd49 : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f3d9306b67944fdb81dad9ceac2c983 : SUCCESS in 16m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bebb8d1e84a348e39f4d5fde338cc8cf : SUCCESS in 30m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/03b5564aaba7498390f50d7094e5d93b : SUCCESS in 18m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7c3f71d2554f4799baa6ba729f8c2346 : SUCCESS in 22m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5610bf5e9c0b4d22bd387cca3bdf18b3 : SUCCESS in 23m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b40e1af3ff74453b0fd1c4fcd0f3af5 : SUCCESS in 6m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3ec8478d22684375bb4fdc986d75cd0a : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/129ad778f99c48bf92aeb45f4a595f4e : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b9f0523fd28243f78e3102f33570ad03 : SUCCESS in 12m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/774006f406a047b0b7cc53d9e233ad0f : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36176c7f0ddd4ff5a70169da7d5f28e8 : SUCCESS in 1h 00m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0cff6659c5934c10b6c89f7a4b799991 : SUCCESS in 1h 26m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce0982a72c144f9e8f263e9f603a8b55 : SUCCESS in 1h 32m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a4890ccc32d4f1899e3f7db7b6772e3 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9867f45399a93b1d7aa1f1e38d08c0a69ff2308c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-09 18:39:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed0edf28db779d58993bac89c1f8278497135c43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 18:40:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8490df1978d2a4cc899323c96a6e66b61aa7e95f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:27:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5375c0ba7e5740f49fa645a9194a4b66 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3420f6d584eb43ff85cb2b4fcbe55de5 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2104ebff0f9042a983a714eb4a817ff6 : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b38f41e0a3e94a0e9b8cb55c614f6efa : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4707a76035a479abb4f71bfa31ed912 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc930e32238548889d7d55e9ba1dd842 : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7ea034e11b7b4fdfa358cb0e2e804738 : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3dfb3b22ad904ecdb83f9dc60dc2f237 : SUCCESS in 10m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/87ab4e56ae904f0eac8e7766e2d23b5f : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/28b21bfd2efe4289868390c23d6a9029 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3476e52a128a45c7bcf60109afb79780 : SUCCESS in 12m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9b755f9b8b644ff39d2418c28d1f7311 : SUCCESS in 19m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5e9557b0cea443daa4747616220e3b6a : SUCCESS in 24m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2065005cd2848b7a6ee758b9fd66b40 : SUCCESS in 12m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4f12ff7cff1c4de0b00188d977b346f4 : SUCCESS in 14m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e33d58659b449e1a38578d2efb07a1e : SUCCESS in 22m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8e91bd3db0ce4e1488c2c8f9e3fd36d0 : SUCCESS in 6m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/58b8c5e930f14c9fb5261bf559fe93b5 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be6a27d04b3248399fab9c9a73f9dbe7 : SUCCESS in 7m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/518d7c8f34a040189cd4106d8cbfc9da : SUCCESS in 11m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87b611b67fa1473e96e950fd7217fb51 : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb454ab269404df8b952e37b3ab04da1 : SUCCESS in 51m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/08736749d2194a86ae81048332e7c8cb : SUCCESS in 1h 34m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0734979527af4d73bc60e68b53ad0ed6 : SUCCESS in 1h 25m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/52d6f96a9fee49bc933b6c7d8589fe1f : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e52ea2ac5118d68030238e45c98158cb1cc6e8d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:27:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d237a1c638675607ca864911826f1cbfc3d34381","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:30:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d9096c6dca470ab82c835c07ca5132 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2adf7a5392f301e9a1757c32729d7c40fdc28762","revisions":{"2adf7a5392f301e9a1757c32729d7c40fdc28762":{"kind":"REWORK","_number":1,"created":"2020-04-09 11:08:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/718664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/718664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/718664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/718664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/718664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/718664/1"}}},"commit":{"parents":[{"commit":"d51c30c84a9e5ce9356fdd390f469d663c333375","subject":"Merge \"update constraint for python-vitrageclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d51c30c84a9e5ce9356fdd390f469d663c333375"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 11:08:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 11:08:43.000000000","tz":0},"subject":"update constraint for python-qinlingclient to new release 5.0.1","message":"update constraint for python-qinlingclient to new release 5.0.1\n\nChange-Id: Ia9db5f04b4c6738bd25de11ebc46af85e51aedd6\nmeta:version: 5.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I35889d77ac855910274a96379f0e7f70cd2bab66\nmeta:release:Code-Review+1: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2adf7a5392f301e9a1757c32729d7c40fdc28762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2adf7a5392f301e9a1757c32729d7c40fdc28762"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
