)]}'
{"id":"openstack%2Frequirements~718665","triplet_id":"openstack%2Frequirements~master~I957892bf0e3cd4a2bc3379466c5c0e437ea9d167","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I957892bf0e3cd4a2bc3379466c5c0e437ea9d167","subject":"update constraint for python-searchlightclient to new release 2.0.1","status":"MERGED","created":"2020-04-09 11:09:37.000000000","updated":"2020-04-10 00:33:26.000000000","submitted":"2020-04-10 00:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718665-1586478673432-fe1485d3","meta_rev_id":"7da27635a8e10aa4933bda01561362658fbf4db9","_number":718665,"virtual_id_number":718665,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 00:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-09 22:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-09 22:50:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 22:50:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 00:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"899dbf79a9be528f49639be546e150df2994375a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 11:09:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a535b0308e0931583841e1994c188fb93e1a9aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 16:55:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71637ce298d4abe9592e67a9c8962c0 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538839950c8e43fd98dc54cb9ff47fd9 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74ef34eea7fc4abab3b3bffeb9fe9414 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f69408bf21d94c98ac4c7ffd2fb7d25a : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbd4e6d1a94b42f3995cd669293ef43b : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e9c2750a049461e87933f1fb949a051 : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a07fc0b663cd4c4f968ecc175a4cdc33 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b2f1d146570a43c0b0a1d14b8a566ea3 : SUCCESS in 10m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/30666bd0722b4a75b91f0948baf8956f : SUCCESS in 11m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/428c878c316d45f8bd7b7d07431840c8 : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/140b3fb3190f42c39fc1ccc0ad028bb5 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ffe6a2daf30140618653964c2e4bc5c6 : SUCCESS in 12m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d6d2ea4914714c3cb8ff12e4bbb7b443 : SUCCESS in 14m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9f01511b5cef40e89cb9c1a499d04354 : SUCCESS in 27m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f9f5f8311cd4bbda2df33ae83fdce7d : SUCCESS in 14m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6d32a7a25ac84de0b944ec73b079153a : SUCCESS in 18m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c28f5108d26441eeb02f722c17854863 : SUCCESS in 19m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/00925fab95bd4a8d89f60ce1b414da5d : SUCCESS in 5m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/581bec4e1ae54fa79dfc9350e50f0929 : SUCCESS in 8m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5848e0460aa4f7485a8b3f2a66c7021 : SUCCESS in 7m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/615cffce654942ea9f8188af2b3f5932 : SUCCESS in 12m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a58d82531c6440d08af7f7ff6e32b728 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b84244301a8542c2a5ddf8caf0268324 : SUCCESS in 1h 03m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7495670a907943b9934d43aa9554bee3 : FAILURE in 59m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0423f49868bf468598aa6a158966811a : SUCCESS in 1h 32m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fbb42ae52f254aa59751ffc829dc481c : SUCCESS in 1h 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"69a5155010aff0cb0e0d9b3fbf192855011417c0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-09 18:40:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5491db14dbbf587fa51a24b5f101a983bbcb06a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 22:00:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88735d1c456b475685c6fbb15908f0cb : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2a2e2271ae42899dcee79119c7d3ae : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a576495fccdd4b79bd9328ed2bd340d8 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c862d4e881884e6db15bfddf80a2f2e3 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93ff08955be6423fb3f671ff4c96c5b5 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eac0c162db9b4996ab4943c368e2a362 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbdf11e42ba744aa914ab8a51f2e09fc : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9f552d39db534421b307a16a1d1c517f : SUCCESS in 8m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/71d8847a566a4cb182743e1893e1b5f1 : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c796d07c4aad4cf4a02a3fe25a200741 : SUCCESS in 7m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/204ff2b1cb574796876bc4287d285d58 : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53e503f1f5164387a8e0b2203bd722f6 : SUCCESS in 8m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/66c5af8af32f4823a174a76e105a334b : SUCCESS in 24m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/feb89556c9b74913890becebac5dc63f : SUCCESS in 25m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c79ebc7e4f2f44a9a94681e005b80eb2 : SUCCESS in 12m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fe33f00a9f414824ac0dc2d47be4b093 : SUCCESS in 16m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e0e36ffe8444a2bbff5f6759b943a06 : SUCCESS in 18m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c50981a4b9974d59bb0692fc9a668438 : SUCCESS in 6m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6b3dab53480142aabd7f317ba4faefba : SUCCESS in 7m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0edead81d16d4fe1979eecad76f34f9f : SUCCESS in 7m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f75e7e11044f4d03b940b516fc3b4bbf : SUCCESS in 13m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a0c0731da97449439812f1075a6938ed : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb235a7448594e03ac8a4bdfe7224308 : SUCCESS in 50m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b183fc549e604b9b8f4ac80695313d46 : SUCCESS in 1h 20m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd739d5cf1154d778852d39d53609e5e : SUCCESS in 1h 44m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2ae5d075e65c49c1a3c986d5af6f225c : SUCCESS in 1h 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"77fce45405e79d70e29e56188071e0d425eb17fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-09 22:50:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"303467dc4c43121d9d496e7b9eb685a4e10ae611","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 22:51:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"938c504c7c266b51050d71dc0477841c485681bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:31:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16fe91011754e2d9fb885b68c8fbf66 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116143b2decc4b90b0937b1318616072 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5163c6760604c5995330ffe729dcb2e : SUCCESS in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4615124f99634b4fa9d3e6a97c430df5 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70cfd752370d4c86811b256922ec71c8 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1670a990364940829bb2a49d767c8669 : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b42dae87dd24299b8ae34f4579f8ed7 : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c9def70b1bc04092958ee3e210b0e6b1 : SUCCESS in 13m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/991367a049f04e1db65a1579b93520aa : SUCCESS in 7m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/68d4ba7b83f747389b99f4ff5e8afe89 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb88f495176b44108b7cd9c3b5baab33 : SUCCESS in 8m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b78a288d5b3140c39b0f30b72a1b29f7 : SUCCESS in 15m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb770c7f5eab49a4ab0e8aac818046e0 : SUCCESS in 26m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8138d06a14924008bdaac71f7e5ef471 : SUCCESS in 13m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f4398725d6f3451a91129424be10b3e5 : SUCCESS in 17m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89cc5400d2694aabb8c217930497cdf8 : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4405fc0ed8544616a549426adbc03e0b : SUCCESS in 5m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0182c12e3ab54f398c8d48fc84dfd8fc : SUCCESS in 8m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/47682173258840c789a365588cdb6ba8 : SUCCESS in 10m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7889ea083d59439f952ea0cd0d1975ae : SUCCESS in 12m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b2244f35428343bba8e06fe6892abd3e : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6b825195e90412b9bd52c0183f7e75c : SUCCESS in 1h 02m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd8af0b674984abd955f2751fb9bc9f1 : SUCCESS in 1h 25m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7d9c574cd5c47b785ca2ddfd06225a2 : SUCCESS in 1h 36m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/650cf0672b0f4207826c422cbddac0b9 : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"38cd60b2c62be706b8d55d888a90865bdafc0f39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:31:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f2d9ce07359aea8a0777332111979ff4e3b73542","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:33:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/800552144fda4fb08b7e4a86c1978e3a : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ed58315005134426226e51d7ccad0c9fbc894c0","revisions":{"2ed58315005134426226e51d7ccad0c9fbc894c0":{"kind":"REWORK","_number":1,"created":"2020-04-09 11:09:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/718665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/718665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/718665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/718665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/718665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/718665/1"}}},"commit":{"parents":[{"commit":"d51c30c84a9e5ce9356fdd390f469d663c333375","subject":"Merge \"update constraint for python-vitrageclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d51c30c84a9e5ce9356fdd390f469d663c333375"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 11:09:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 11:09:35.000000000","tz":0},"subject":"update constraint for python-searchlightclient to new release 2.0.1","message":"update constraint for python-searchlightclient to new release 2.0.1\n\nChange-Id: I957892bf0e3cd4a2bc3379466c5c0e437ea9d167\nmeta:version: 2.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ia40873a4bc730e34ad3b22803bc94c012f70af10\nmeta:release:Code-Review+1: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\nmeta:release:Code-Review+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ed58315005134426226e51d7ccad0c9fbc894c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ed58315005134426226e51d7ccad0c9fbc894c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
