)]}'
{"id":"openstack%2Frequirements~718666","triplet_id":"openstack%2Frequirements~master~I012c2e251c792fbfbe06806f8965f084c977d387","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I012c2e251c792fbfbe06806f8965f084c977d387","subject":"update constraint for python-congressclient to new release 2.0.1","status":"MERGED","created":"2020-04-09 11:09:41.000000000","updated":"2020-04-09 20:27:17.000000000","submitted":"2020-04-09 20:27:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718666-1586464037188-a4cc011a","meta_rev_id":"4376afff3a8ea6b185f16fc8e22ebc8589cad01c","_number":718666,"virtual_id_number":718666,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 20:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-09 18:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-09 18:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 18:39:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-09 20:27:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98ecc432129dc42002f276282816e92d119c3746","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 11:09:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b396c119210b3f89432bf118d38512a03c0315cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 16:38:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21da15acdb4452396cf6fb10a88d55d : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50c8b97b1674374a7e751f4a3b6c1c3 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de09ffaace404088a10ef91510888650 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/401d47b0d4ad44309006d89adf9b0921 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1394570a6234f3eadfa47c1290b8378 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe3fce7f0f4a4da095a0a91b130c18b3 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23d870ad32534e18b1dace5ce5efe4b0 : SUCCESS in 7m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/137468bd92994adfa0c7e740e65e95f9 : SUCCESS in 12m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c0650ab76ad42cfa0cc96fdc469bfa0 : SUCCESS in 12m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cd0490f51fac4e1a85632f0bac6c691b : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad7b735d108942f98e285f5950c6e3c2 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd8d9902c94b4da1b6f9049a68d95dfb : SUCCESS in 10m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7d65d28ad3d84cc1902aebd7b79b2c8d : SUCCESS in 22m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/81c0a855788f4acea843217bef7f6714 : SUCCESS in 23m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dd5b8ee4f233416c9b2937079eb4033c : SUCCESS in 16m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6aab7490e8df45fa944fa18bcdbd013c : SUCCESS in 16m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7867c8143a914119961a623709fd3002 : SUCCESS in 17m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/865f98ad2bbe47919f72bc7c6fdc5155 : SUCCESS in 7m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5ed601be90374ef5ac8d2ffa98e2be6f : SUCCESS in 10m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49dcf3a2c1a543fda58a21da697f0472 : SUCCESS in 11m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5dad6be1d74c41aa8d368de9aa464de8 : SUCCESS in 11m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/064fc350f9634cd9bc76f9bd5d14eeca : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/676df96e663444b5aa058ad08c8bc408 : SUCCESS in 55m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89aa31391a1346e0be37908bd92dbe74 : SUCCESS in 1h 09m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8815b1afbd6e4e368bc90796354be18a : SUCCESS in 1h 14m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4d2409ff9b174fe9ac2611c4cf5fc06f : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":1},{"id":"07588299447c9f202487dead3f7187920fb38494","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-09 18:39:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e0152a21d6212d0779713e64622673bcc62155b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 18:40:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"509a7bbba29b031aefa682f623a2d96c94c9ac5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:27:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b82ea3d02f4ddeaab8d5f66739bb36 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d4da20b97db4c09b75d791ed1343a23 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93b6c3d394ea4111b184136cde580533 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a05cac855fe440d0bc41fe53b68f8024 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94ac88b2f41645c1b88b0a960142b856 : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61a6d00c087744898a52e783969efb28 : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/572c8e8a996b4893aa5702724965543b : SUCCESS in 10m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/691ee36d60554c99a9fcbc3a515c35e5 : SUCCESS in 11m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6e36df2359ea46a3adfa01c1fad4f4d0 : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/293adb89c7304966af8456282cc2dffc : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df60bcd29dda49519f4ae370fe06a58a : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5058d9d1399745ebbc8567edac18ac0a : SUCCESS in 15m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a4aba372bca4195abbadee060b9fa25 : SUCCESS in 28m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4701454af04c4b0b9a46ba5c83bc0da8 : SUCCESS in 14m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d6052b4a2cdb40f597096193747e76de : SUCCESS in 15m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0560869ff8f640cc8bcad79860856138 : SUCCESS in 17m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65183784bd544089a5230b9f8526b881 : SUCCESS in 6m 35s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ac421fe3139b47a5b69b145e86141b1b : SUCCESS in 9m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ffafff66408944a5b0007bde581f1deb : SUCCESS in 7m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0d7a65b5906541bfbc6a31eca5a21c3d : SUCCESS in 11m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/df7f014950dd4783add484825ed38bcf : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6f8e10b6a24409b9277047832343ba2 : SUCCESS in 50m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e540dcaeb494b129e05237b33bd2046 : SUCCESS in 1h 12m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb5f5315fc344c4786bef3b2cbbad5b7 : SUCCESS in 1h 08m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4be6c7db354146739b4dcd2c4751ac8f : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":1},{"id":"671a06b4d41f248a0fc22a2f76c378a5ba50b6f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:27:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f613c694ae42ad05a0dde80eb166193c0e13104","revisions":{"6f613c694ae42ad05a0dde80eb166193c0e13104":{"kind":"REWORK","_number":1,"created":"2020-04-09 11:09:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/718666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/718666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/718666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/718666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/718666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/718666/1"}}},"commit":{"parents":[{"commit":"d51c30c84a9e5ce9356fdd390f469d663c333375","subject":"Merge \"update constraint for python-vitrageclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d51c30c84a9e5ce9356fdd390f469d663c333375"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 11:09:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 11:09:38.000000000","tz":0},"subject":"update constraint for python-congressclient to new release 2.0.1","message":"update constraint for python-congressclient to new release 2.0.1\n\nChange-Id: I012c2e251c792fbfbe06806f8965f084c977d387\nmeta:version: 2.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ia8a3fdcce4b4865d7ba787f5fe322e995efc8a75\nmeta:release:Code-Review+1: Eric Kao \u003cekcs.openstack@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f613c694ae42ad05a0dde80eb166193c0e13104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f613c694ae42ad05a0dde80eb166193c0e13104"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
