)]}'
{"id":"openstack%2Frequirements~718668","triplet_id":"openstack%2Frequirements~master~I8a68a1c7c05925af69a4027aa148933727d5db88","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8a68a1c7c05925af69a4027aa148933727d5db88","subject":"update constraint for python-mistralclient to new release 4.0.1","status":"MERGED","created":"2020-04-09 11:13:17.000000000","updated":"2020-04-09 20:34:01.000000000","submitted":"2020-04-09 20:27:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718668-1586464061177-8cd73cf9","meta_rev_id":"9c8d2802860c0e4bff993ef037a4501ea1f9c558","_number":718668,"virtual_id_number":718668,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 20:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-09 18:39:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-09 18:39:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 18:39:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-09 20:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37ccea181eae9d91a15698744984378473381cdb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 11:13:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75737ccad4601ebe69f98bc93b052cc37d322e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 17:19:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f0f3b58c79455a8114bdd8dea79ef7 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de303738dea24a969617d6800cf38c11 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15a6ba84631d48efb9f64a57257a9d86 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/217aad7b81f745fdad86d592716ad537 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c1150597ce049ab92af96569c988e87 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f2cb70bdc674ccd8f16ccb7cc53b5fb : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/084c3552f8b14791acbedb5209520ada : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8548d430f417447b871376968ff8055c : SUCCESS in 9m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/75b8883bdd004820b43d8487ff3270b0 : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a8133dfed02e452792258a0a397902c4 : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e47d0f8f43244058f19fab029c94bd2 : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab34ccd5920a4fd3ae352fc7c2aa6def : SUCCESS in 8m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f616b3226b9346fbba758f475ed14954 : SUCCESS in 22m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e4c95d7eeb9a4348969bb2504a1d09af : SUCCESS in 28m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/50d3520bdb5d4d32b28607f8c17ad1d2 : SUCCESS in 14m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e38abde58dfc4ee7bc16a9e2fc347466 : SUCCESS in 15m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/843348efbe1346588b5c1f0df683e3f2 : SUCCESS in 18m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/00808f2785c6410093e4c7cf8bde762c : SUCCESS in 5m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5c337034465249eb8fa2fa19b41c23d1 : SUCCESS in 7m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a40c3c75b614598ac71331b0df6684e : SUCCESS in 9m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ef811977228e4813a113f99bebec45a3 : SUCCESS in 14m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6256ba23754c4b878abad598f1936a99 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57abcb0147454adba14975ed6cc86230 : SUCCESS in 1h 00m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/807f4af5336a46b0a0d1fde5d4222033 : SUCCESS in 1h 23m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5e765998f004e77af6023be1a2d84c5 : SUCCESS in 1h 28m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b68703b8181e4a58a2424ffcb8755341 : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dd03c337daff543bd6479037b0bc8372e5ec2d3f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-09 18:39:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1921e6040891e9b69890d5f7faa5b158a1485870","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 18:41:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfae6b8b110a194656cfe563df371ba75ec8ceda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:27:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/219f3c1fd00a49ae8b6f9ffc6ee30f65 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd05d7662a44b4d992a41fa80730308 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fed44b931b6409d9682dbaec17435a7 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca9428cc08c74eb8a574f7c49e53a831 : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08022c7db50341398ef5c1eeddb4432e : SUCCESS in 6m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5ad2378be4741e5ab2d733d1bb1a8cc : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/265cab34465e4580b7b73f4ecf686127 : SUCCESS in 13m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9359eca0efe94b9395db3a297c2d537e : SUCCESS in 10m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e3c84819e2564d0eaa5e1c0a065fa840 : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5cd67e85bb28417c8278af59b4699708 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2384094435f491e8bbcc708dde3f4f0 : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/30f632f44a63471ca62444983a050361 : SUCCESS in 21m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/19b79c0240964711a207599ddcf9dbd7 : SUCCESS in 28m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6c7e754be41746fb8eab726e96ce6d2e : SUCCESS in 15m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ab6b6a8471b24835a69ce37e3131ef03 : SUCCESS in 16m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/967ba7f4723740ac8206adfab3f4bfd0 : SUCCESS in 20m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/971d8633e6bf4999ae37429f878fa27b : SUCCESS in 7m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/29dd76a12a6245f5be2f93a26b09866f : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2c6b423f9fc46cfaf05613d4a4882af : SUCCESS in 7m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f321cc8737944190a45a6742f3b815df : SUCCESS in 11m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e44bf0dc6fc64da7b90b0b94ed5a85e0 : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3cb6a44046f4ee1ae15e055a2563379 : SUCCESS in 51m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9b7ba2249d54334937cabe6c676b159 : SUCCESS in 1h 30m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db420714df544dbdb811982451cc07a5 : SUCCESS in 1h 21m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bf6172e3c23c4089ae1ea74640c209ab : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7f5e5515bdf193613497db820c053d03866c3697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:27:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8df288a6808044c629e3521ec0099da74ce81910","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:34:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e5ec1967f447ada1183d58495072fb : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca597c4a972fca29f49d0fa5ad72dea3636283dc","revisions":{"ca597c4a972fca29f49d0fa5ad72dea3636283dc":{"kind":"REWORK","_number":1,"created":"2020-04-09 11:13:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/718668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/718668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/718668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/718668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/718668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/718668/1"}}},"commit":{"parents":[{"commit":"d51c30c84a9e5ce9356fdd390f469d663c333375","subject":"Merge \"update constraint for python-vitrageclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d51c30c84a9e5ce9356fdd390f469d663c333375"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 11:13:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 11:13:14.000000000","tz":0},"subject":"update constraint for python-mistralclient to new release 4.0.1","message":"update constraint for python-mistralclient to new release 4.0.1\n\nChange-Id: I8a68a1c7c05925af69a4027aa148933727d5db88\nmeta:version: 4.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I6be753a234b68f0f6be5df86bdf5e7a46bef3183\nmeta:release:Code-Review+1: Renat Akhmerov \u003crenat.akhmerov@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca597c4a972fca29f49d0fa5ad72dea3636283dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca597c4a972fca29f49d0fa5ad72dea3636283dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
