)]}'
{"id":"openstack%2Frequirements~718797","triplet_id":"openstack%2Frequirements~master~If45debc512e5ba9472be0d326c8f821d69e1956e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If45debc512e5ba9472be0d326c8f821d69e1956e","subject":"update constraint for python-solumclient to new release 3.1.0","status":"MERGED","created":"2020-04-09 21:05:17.000000000","updated":"2020-04-10 01:32:34.000000000","submitted":"2020-04-10 01:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718797-1586482217294-887b7afd","meta_rev_id":"25cf42decb726c64bcb11711dad53fd31d338586","_number":718797,"virtual_id_number":718797,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 01:30:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-09 22:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-09 22:50:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 22:50:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 01:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db4279fc10f4f8132345787b100b4f39e9af7e3d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:05:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5d0d4b290e48667cb951b6fabf5558e81c66a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 22:41:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca991e82f4a54d00800fe9d8c72c4cb8 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8375e44be26b47a683e1fac163c4f203 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c52daa49da12402c84890ffb58db7ea5 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c44770016a814f5fba6936aeccbc1614 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c790a37b4814023ab1ce603b9383166 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3be7bd6842154a42a39bbc9c512fe6d6 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94917df8028648ad90bc4dc3325536ca : SUCCESS in 5m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/52630515d498436ba515f9908c0cc523 : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3e4e3c7da0ba49a58d7c86aa72ff1bc9 : SUCCESS in 12m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e1396a857619440984575ae2e7e19c90 : SUCCESS in 7m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3bdd9e032884412e9d061ee9239fc9b0 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f17bb9548963476997460497ad59db3c : SUCCESS in 11m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f5233a5163f4b11a0752942b47ea1f4 : SUCCESS in 16m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/88407aa85bd04f86b2a73b7142934a1d : SUCCESS in 27m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/548165226f9f400f99277533ede46656 : SUCCESS in 13m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f09554cb46954e859102ed0983e23914 : SUCCESS in 16m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/272858273c0245cfb2a0409006cc7990 : SUCCESS in 20m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/77efe68d27bb4e0a86f175e89e810972 : SUCCESS in 6m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/39c516f0276b427b90829c27cda8ebef : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f417442758b340099dd719f2a9e5f6a0 : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5042dda89a4347ed85598b16c2c12201 : SUCCESS in 10m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d9c418cfeed24f41805e6d928f475812 : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bc277c919944270a81e132bcf2e8e63 : SUCCESS in 53m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e28ffd5e5d614cecaf748ba722ab735b : SUCCESS in 1h 14m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28bf504b24b44328baa2467727575b78 : SUCCESS in 1h 26m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6adf88e0ed4947c4bba9d90a3eb155d0 : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"28f9cec401abe77813385fc2a7f0bb72cc2c2aeb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-09 22:50:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"795d3bec1db59cbce67f0c8bed0efa60209ec407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 22:51:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8548e1071642524c4ff9f9fef3668addd98f84d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 01:30:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08fc2413e02466fadd6a4e3d945b284 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b32eaa0335b41aab95d5e6fcba37a42 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2632a1e53a144e9abf851e1bb15a0c1 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a209ac9b618c4e9e962be05999314569 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d70e44f957f441f97903d9635e0b1b7 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e776da7336da4881b306fc3230ed3863 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c5c5d165995e4897af54097f4fb68af9 : SUCCESS in 12m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/92c67f9d9383459484f8f2f307c0eacb : SUCCESS in 11m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0d831622fbd44ee189924dd2ef490ca3 : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f476c05861c4fc3bb59e75be90c9313 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/342a69a937ae473a994ee87da40ca96f : SUCCESS in 8m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b7dbf15b47e46cf8a8a15b6caeb8304 : SUCCESS in 16m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0742da54784045d8bff90598459ce6e4 : SUCCESS in 28m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f86eb6b4033749b4935010c3a2c1cab2 : SUCCESS in 13m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ab9943c80dbf404d90d84374824d0010 : SUCCESS in 21m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55bc5c636a6b4167a6cd42e3ee37e929 : SUCCESS in 20m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5147cb9238734a7d85a33fca681f404a : SUCCESS in 6m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6e52bd74dc5c42b5b3c8523a5458c737 : SUCCESS in 8m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fceff27eaf224183ba4030204ad0e6da : SUCCESS in 8m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/124dc14df7bd4877a315b5471d5ff10d : SUCCESS in 10m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/343c047f09764305a15c5473675d0ad2 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b86f06512af4b92b00a547e2d0eed16 : SUCCESS in 51m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/943c9d39f4db40fb9fcf94ec5fc8acc5 : SUCCESS in 1h 18m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e9039033672433898411fdfffa4954c : SUCCESS in 1h 28m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/963a0783a5a448b68d0da1f458d16445 : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e10ab71f5fbd354786ed72d57fffd49259b706ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 01:30:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a2376a0a38efb8c436e8433e634d8402a5958ef1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 01:32:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd8f85e77ae4af1a97decc5579d9d91 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30773d1666b811b1a0c93cfa9fe05b1c3b936cf9","revisions":{"30773d1666b811b1a0c93cfa9fe05b1c3b936cf9":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:05:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/718797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/718797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/718797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/718797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/718797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/718797/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:05:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:05:14.000000000","tz":0},"subject":"update constraint for python-solumclient to new release 3.1.0","message":"update constraint for python-solumclient to new release 3.1.0\n\nChange-Id: If45debc512e5ba9472be0d326c8f821d69e1956e\nmeta:version: 3.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ib0e935d7f2b484b5808ba385159661d0a9d6f78c\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30773d1666b811b1a0c93cfa9fe05b1c3b936cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30773d1666b811b1a0c93cfa9fe05b1c3b936cf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
