)]}'
{"id":"openstack%2Frequirements~718800","triplet_id":"openstack%2Frequirements~master~I4a99e4e4f9ef77f6588a92d7680e6a87c0f6d128","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4a99e4e4f9ef77f6588a92d7680e6a87c0f6d128","subject":"update constraint for os-net-config to new release 12.2.0","status":"MERGED","created":"2020-04-09 21:10:40.000000000","updated":"2020-04-10 08:26:59.000000000","submitted":"2020-04-10 08:24:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718800-1586507089885-a4cc011a","meta_rev_id":"b7c0ebfcdd1ca6219a850e2b816f980777063580","_number":718800,"virtual_id_number":718800,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 08:24:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 23:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 02:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 02:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 23:07:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 02:12:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 08:24:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d3b3474bd641e0333c621619befac784424e7c7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:10:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f652ef315af360682d6063e332a42ded014acc4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 23:05:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045a30c5d22247f9bc35ebdc5f2a13c6 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d2bebc7c7641c4b133800cc0e7b20e : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d06f3081b8a64b8785090402a905ebc2 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d279ccacbf7042bfb9db7e8938695683 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab49ae4fcb694869b547b8d5056a90de : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c57062f6c56049e3805b373b2aa71735 : SUCCESS in 6m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8dcf79323c045cc9ac01e974a6f199e : SUCCESS in 5m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/efb38641a83245ad8ef1b15aa1dc6dfe : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c46f585dee3347909a682f1f53653d94 : SUCCESS in 12m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8400291fc96340fba02af7bcae213053 : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/afc271711e464a17a322020bf1afbfa8 : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61f2c5f3a45146cc83586777fbbd79d0 : SUCCESS in 9m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7d45fc1b663949cca062f5eccd6ed272 : SUCCESS in 15m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1054cf21903d4b52950c5638119a48c3 : SUCCESS in 32m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e27a3d0cf79c4c438fa0d8a667bf1817 : SUCCESS in 14m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3ae3fbeec7ff4c5c910678b17ecc0ec8 : SUCCESS in 17m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c60308aecf77447d8799f6073098eb6d : SUCCESS in 19m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/893d460d7e5b4d8d8835241c1b8986c2 : SUCCESS in 6m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f2299a5af7384e6e8bf75b3154e956e1 : SUCCESS in 6m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af2e7645157743129a73eb93b20f973f : SUCCESS in 9m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/702c7b7edf034a11aef648d64fca3118 : SUCCESS in 17m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/268abc6d9b264a1487905c83f1467f37 : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/956dc7108ca94a039255b08c8c5bfc66 : SUCCESS in 54m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed30be0677424a93818d1a578e73bf0c : FAILURE in 1h 10m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/089c0d51211e408e9ebe88dc25548bab : SUCCESS in 1h 44m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/52e7de3728a8420988a5cca25daeb736 : SUCCESS in 1h 06m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1325d844ad8381f1e5b63e48ec7ba0968a4eb698","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 23:07:29.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa2df2605552cb888eb143f08e0b61c3dabb7a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:36:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/219793fdbea74ed5b2a9a5771065c44a : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ba77531a15c4d04b725314e99aee136 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bcbb432532440f09ba5a2a6d7eebb9e : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57ab55c99a634242894262ad6f8f2fa6 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95f985fcdeb940e2b7e07ab479b571d3 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48b43b692f7841acac271b6078db8f21 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9eca4716e7c04ce09969cd3bb37e260a : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/faf37705963e474bb293375d3195bff6 : SUCCESS in 7m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/26da77a414044513aea53da934b0bf98 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/58fb85f71d144c7cb8d1d37e0f0a2589 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8a0351f1fe854dc981242832f0587f0c : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/05b914b129194f97b0ba24c958b15b58 : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bfbde67d4b31411b927410a3f2d98b58 : SUCCESS in 24m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e8f8463bb61a4da9b50fd6ceb1ebb361 : SUCCESS in 24m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1cf71b5a2eda4ba4a219ca453f257e31 : SUCCESS in 13m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dea31d53c08c4344b5ac23d04d618989 : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1450678230b24e5db66dea6a7b7019b2 : SUCCESS in 18m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b757a71c344d40ceb50682e34747487f : SUCCESS in 5m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cd100d0167de43f9848b6382cbbed14b : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cd18b10bbf1a45de85be421ab615b9b3 : SUCCESS in 6m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d467763b6db402c8fb60aefad47994c : SUCCESS in 11m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dfef4e338e7440dbb9154739d59699cf : SUCCESS in 30m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1abc76fb601748f3b2d2c0d496c49a89 : SUCCESS in 50m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd0c70299f894b158d817aafb0c70b59 : SUCCESS in 1h 22m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6168ecf2261a4889a00aadc6808793e0 : SUCCESS in 1h 20m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e2c4040a729e48a5988db5d1af47e856 : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4db1ababf79e40adeacddfc39f5e2d54dcd2d2ba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 02:12:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37fa5b43a5c7f614a872162061e1a1fb9b5f61f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 02:13:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7bb2dade280be98fc469aa120cd5affcab7520c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:24:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"67cbb024355051301aa20f0ce0c6dbb2487cc9d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:24:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee33fff49c041ba98274c02a63e4ff1 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c5fefbb3484807b08772f02c548e8f : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a862d818933438890ce4dbbedd17ad5 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1638581ee78e4c33a393f88f482c5e17 : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8ecb7174316450488a90e5c6c0c92bf : SUCCESS in 7m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16a22c7df0d34bb88f8120d28f995604 : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4bd759476aad43f69b89e49a181e68e5 : SUCCESS in 11m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4164068a829b4a688ab515d7bac8096d : SUCCESS in 14m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e5c08e1b581045ce88adf728fa389555 : SUCCESS in 10m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/038d12e9337640d690a395ef96d7f114 : SUCCESS in 13m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/833693089c724031a21672ce699aea6b : SUCCESS in 12m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5806122550ce4072a6fb290119169565 : SUCCESS in 26m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eb793160051d440f9b9de9e5369c9dae : SUCCESS in 30m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37754e740f714a9383a38138c2aee30f : SUCCESS in 13m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/90403722a7a246f498283f79b703ad8e : SUCCESS in 20m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ac9483a33ed4ce68e02c8872c49f5c2 : SUCCESS in 22m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f32b2b9eedce4b0183b3df1faac47657 : SUCCESS in 7m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8eb7ff7d6d3240a9aef4caa7a78a0b05 : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c0ed2bb71d74d93b1904ed6c125c83c : SUCCESS in 9m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b66ae019a4e74206b5e344867ec29d8a : SUCCESS in 17m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/941cecd562204a2cb6a26198466176d5 : SUCCESS in 17m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/263008bb545446f190d2fbb579e3de04 : SUCCESS in 1h 03m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb26f390ea8044e1b162407efe85f136 : SUCCESS in 1h 13m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e75728741f214a638be02ae820f1a656 : SUCCESS in 1h 22m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/147d5c692caa496e8654f333e5d672fd : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d37f031b2ae846155c188db052d57d3888c2b019","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:26:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96b6f38553146a99fba4036b2a0edcd : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c25a7adb0906cac0c9680bf1390413808014303","revisions":{"9c25a7adb0906cac0c9680bf1390413808014303":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:10:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/718800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/718800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/718800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/718800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/718800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/718800/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:10:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:10:37.000000000","tz":0},"subject":"update constraint for os-net-config to new release 12.2.0","message":"update constraint for os-net-config to new release 12.2.0\n\nChange-Id: I4a99e4e4f9ef77f6588a92d7680e6a87c0f6d128\nmeta:version: 12.2.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Change-Id: Ia52230140b5124423ca904fdc1e46f67b6e569bc\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c25a7adb0906cac0c9680bf1390413808014303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c25a7adb0906cac0c9680bf1390413808014303"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
