)]}'
{"id":"openstack%2Frequirements~718801","triplet_id":"openstack%2Frequirements~master~I43fefd95935d5b1f6d9c616401d59ce3882b2c1f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I43fefd95935d5b1f6d9c616401d59ce3882b2c1f","subject":"update constraint for python-ironicclient to new release 4.1.0","status":"MERGED","created":"2020-04-09 21:12:50.000000000","updated":"2020-04-10 08:24:52.000000000","submitted":"2020-04-10 08:24:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718801-1586507092646-30de0cd9","meta_rev_id":"92d5beef909f2036acddd0f3b42c06d0d638d6de","_number":718801,"virtual_id_number":718801,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 08:24:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 23:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 02:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 02:12:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 23:07:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 02:12:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 08:24:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e0dadcb725d92882b8f3fceb3bbb8e358c35a2d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:12:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08fbbca89b8c0c3cc22238cd63f2538d48577e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 22:57:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d35e91f810541648e97eb27e53add87 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e405d3d2fcc84f22beedce4e364844c6 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e2195ca4b624dc1bc600088cca79c40 : SUCCESS in 8m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d94fcfdb2ffa4e13950b94232f5beadf : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d6d573f5b3f419c9351b4692aa2c990 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25ed677f97724784b984bf013438a722 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8687652bd1d544faa489668173a1f27b : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/363d185bbaac478badf05d1b03f04c7b : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/61ea90a1f1c247418b67405286f1037f : SUCCESS in 10m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d04a255d373e4f479aef301d644cec45 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9f11d2dc8801456c8b355fe5b4ccd38f : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52e8ae6723a6408aa955b9e94288bc10 : SUCCESS in 9m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1243e41e4e7a466383beef4186e6031b : SUCCESS in 20m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2d5ed766ec3d4d6fa2d6c0c0577d4cca : SUCCESS in 25m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/62d8441fd98a466788269a746cff1c51 : SUCCESS in 19m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7e7a1770e2fc46a1ac288516effaab60 : SUCCESS in 17m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/332926594cb44c62b70462460dbc59ec : SUCCESS in 17m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d571afa0bbbe4a48a127e2c370134015 : SUCCESS in 5m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fa44107c89514f8b95c33c118e4f2661 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af22c294c54f4a759f53349cb193f5fa : SUCCESS in 9m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d9802e7c1f994989b4992474ef109ab2 : SUCCESS in 13m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dadbe50fb5e64e8eaecc6de335bdfaed : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7162f917913b453ebf77b4dfbd951a24 : SUCCESS in 1h 01m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80ebcb25486e4180bdcc72afc6711413 : FAILURE in 50m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91ca94b47bac4ab79638452b6c4ad63d : SUCCESS in 1h 32m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d90924e9a75546abb7affc99cb169697 : SUCCESS in 1h 13m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6b1667b7be1c87652f691b9d6f5cec38549e326c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 23:07:07.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28af530d9854b7a268dca17ad15d9ce87ac67a39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67168b4bfba34ce2a5694c34ec107801 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c4214723d040afbfe9e42cdf89d222 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e084d859bc724c82bcbc8217fbda362c : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ab50d12e5a448a8865ca82ea31b5a0c : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c166f707ae304996bb4a61e210d9802d : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94dc5d894c1848538f3749ba3ce1439b : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5617d2a5c1c042b1b3d56eb8600f62c5 : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/086227fd7e3a4451a65f5dedd116f589 : SUCCESS in 8m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/abbf5e2e7034438eb0d3407bf76f2b4c : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ac0d5ced8984716bc1f4a5c2529489d : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f06aa54020984bc3814dc64fb4a89ee2 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/376a6c1aef0842139dd4c5806f61aedd : SUCCESS in 8m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4d7e1e685d4a4d1ea564d003198a91e3 : SUCCESS in 15m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6966ab742db1429baba329d0008f4403 : SUCCESS in 28m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b49ad4347c734380b74e88ce5f600def : SUCCESS in 11m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/19787ccfbe9f4355858fba155373519f : SUCCESS in 15m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f75f29023aca49e696a723162f9e6d25 : SUCCESS in 20m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c604275d5502499b98809ac5d001a9a6 : SUCCESS in 5m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9378c19cfe5c4a7d8252d99cff678bf1 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/009573669ba84cf6972f3c8c4ddcc387 : SUCCESS in 7m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/090c795f29224eaa8159eef932436be8 : SUCCESS in 9m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fcdec1cc3ef34a7db4eafec2e75f5147 : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c01cdcb750384e88aa782fa93ad42e27 : SUCCESS in 51m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/453d11b0c9b74e9591c9bd67363f6bb9 : SUCCESS in 1h 11m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5143351acb44fca9fe33d2b0f51d7ef : SUCCESS in 1h 20m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/89f78dce4975403eb5240acf42fda8b2 : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"54c627891a990e04fa22eeae177abf64fa279b1b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 02:12:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e50e2e82423562c6e76974e2350e34c3e72cf02d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 02:13:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c3e25b80cba6a0d36b906464e1ab7a0dac99aa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:24:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9230d565d666408c8fa1cc4f61ec78ac : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63eb6f6fe4934e4d8fbbc7e8ab4a9ab6 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0893eb72244046adacb696d64e000562 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f64c40c6f2de4902a0a56c992d2e90e8 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d1f48b9373a4a95ba68f456ee62c827 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e7bab697fe943e495c45e0d3224f28d : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f7185d55df9941e3857364c3c4dc8de6 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b17770a703e14e7984dd0ccd8b44b859 : SUCCESS in 10m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ba187d92a6184b7eb349605c2ab92587 : SUCCESS in 7m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/24117fa7982049f08859541301e30371 : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b60fd8ed2fe2405ebdecc2d06e7288c8 : SUCCESS in 11m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9dc74a49098c459cb233718f6aade861 : SUCCESS in 14m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4473beb9fe864ff99285b1cacd7988c2 : SUCCESS in 28m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb3d3e1e9f354fe990c5ae55f9d5e614 : SUCCESS in 14m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/87c8f9b121c54d05afeb9fcee0c62680 : SUCCESS in 18m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc162c96439c4f15a53e2b9dff9e0098 : SUCCESS in 17m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/57df9ad383f44953ae31e5ca950afe20 : SUCCESS in 7m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/81633957b30f4e3b956e6a62d188dda2 : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ff4c0cd998f4c2997a251ed63883eab : SUCCESS in 8m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1cdfbb4539a94722bc1cbaded4ddbfea : SUCCESS in 13m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7e42cff857964e65b9535c9f1b077b13 : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/496e0faa38474ed5908c609b650abd47 : SUCCESS in 56m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/05afee4a51be4939bac6c86f1bc16a1c : SUCCESS in 1h 17m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bba62543f2f4d90908f34ecdfd4ca71 : SUCCESS in 1h 19m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5224cdc5dcea41f7b39f73df781608ec : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8147e3ed1c581cbd801f67f2d7a63bd861476545","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:24:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0bea14176f9aee62109350d4d4e6d355d2bf4f3","revisions":{"c0bea14176f9aee62109350d4d4e6d355d2bf4f3":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:12:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/718801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/718801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/718801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/718801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/718801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/718801/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:12:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:12:49.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 4.1.0","message":"update constraint for python-ironicclient to new release 4.1.0\n\nChange-Id: I43fefd95935d5b1f6d9c616401d59ce3882b2c1f\nmeta:version: 4.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: I083c9ff1bef0b3178ba17046745579e3bedb90a6\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0bea14176f9aee62109350d4d4e6d355d2bf4f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0bea14176f9aee62109350d4d4e6d355d2bf4f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
