)]}'
{"id":"openstack%2Frequirements~718802","triplet_id":"openstack%2Frequirements~master~If139ec2d20db213308620deb233ef393ceeea367","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If139ec2d20db213308620deb233ef393ceeea367","subject":"update constraint for XStatic-Hogan to new release 2.0.0.3","status":"MERGED","created":"2020-04-09 21:13:03.000000000","updated":"2020-04-10 08:24:51.000000000","submitted":"2020-04-10 08:24:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718802-1586507091215-a17bf472","meta_rev_id":"48ff8a6bd345477936fa27a59521211b993b2211","_number":718802,"virtual_id_number":718802,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 08:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 23:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 02:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 02:12:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 23:07:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 02:12:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 08:24:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de1c9419a440d8ec4020dcb3a247440346f5158c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:13:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82f32a4ed46ccd20cfd39fdb0a195d717bc9fc67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 22:59:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ba576e2b7d44db9d271554e24049d6 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb38ac2c28c14630a0c3065aa2399159 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d35e5e5635e486591b353763a90051f : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb4b4459737b4a0299f76811e2217c26 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c594b7e8338468382772aca2160d38c : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/231b61d4e44b412ba31c94bc279e7c2c : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e3ee237c62041eb92fe7384685bc552 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37709e3670074a119866f0b34cce04b2 : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/153188b543834cd4849650224ff5dca2 : SUCCESS in 14m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5297d4add20c450787267bb85d2437ae : SUCCESS in 7m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66f9623efa124e708afdeda5590e508b : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e867995768b0437c8ca56b60560c543f : SUCCESS in 9m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5f50bdc67c1243c59b1b07513fec22e9 : SUCCESS in 17m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/490d772ee9d7420e96d5ac2277a81b2b : SUCCESS in 33m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1a3d1192767c4616807d15955008f60e : SUCCESS in 15m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6e9fc32883f54a13813641fbac0ac44e : SUCCESS in 17m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf171a759b2c47f7b0ba70575ceafcab : SUCCESS in 21m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/333c0ed2540649b6beb96e27d14b4540 : SUCCESS in 8m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f1fd1ca8e24a4b668cde2f827076cbbc : SUCCESS in 8m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55610f84459f4c98ab0a63ea8e2cca3e : SUCCESS in 8m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b3b2d20a1c9540dc8ff0d38cd89a9ccc : SUCCESS in 12m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d314f09624ac4755a90eaecefd3f9a2f : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1b1a770ac31400285695807c2a5d117 : SUCCESS in 1h 01m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/338b9be1cc5945e1af07d2309b5637e4 : SUCCESS in 1h 31m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f942a3e24b67421da7ab9cf8196a28d8 : SUCCESS in 1h 13m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c803ddc006b34896a925d453a16df984 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4747342c7dae8844760fface2c658502c5b94da3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 23:07:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61c01518d20fcf26a58807720e29b677011a93d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 02:12:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b0b9521e9ee7488eaf999da4938d4b6a2d000a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 02:13:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e293cc335ab0433c11eb1a029b6c75f651a90e7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:24:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b051eba1e3ef8aa5c1670d9b5b5c001203f71a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:24:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a083b22ac0f45d187abc30f230bf08c : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029e54bf073a4a00a1de14c3487ffe5b : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/053b7dee6b12496da09a2875ce3fef81 : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/479c0ff3ba5347cfae4a8f7f0f6239ee : SUCCESS in 8m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa81fdebce724c1f87416a8bfb1858be : SUCCESS in 7m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c9b292a6fd1465bb6aee8623d943bc8 : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6017d47a8625450caf195e7bbf179c12 : SUCCESS in 8m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9b9f4b9057b740c598aaeeca45685312 : SUCCESS in 15m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4cf23d9acc424b7985095dc215b72657 : SUCCESS in 8m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2520434cc74a4429b56bbcd1dc2afe63 : SUCCESS in 14m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/530d611e6c5043b5bafb79956c719804 : SUCCESS in 11m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fed30a5a0b304adf92282a6de2f1a45d : SUCCESS in 17m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8bd2372d377f43d98644a025980c00f1 : SUCCESS in 30m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/80f6c5bddfdf42a892a720f026dc55b4 : SUCCESS in 13m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c410600160484bd4ac086e3a6931ed4a : SUCCESS in 17m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c5d6a9e756b49e9b5a03149c21da06e : SUCCESS in 17m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f5eb48a0d9eb4bdab19c6d2cbc685bdb : SUCCESS in 7m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dc49f84eef1e4f188d299c2355cf07df : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6669087a6b954af9b3c703576112202b : SUCCESS in 8m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c455427f6b174968a4cae3318ccb2710 : SUCCESS in 12m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/00b77b0e4acd493e990aed191e2deccc : SUCCESS in 14m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38c495672ae64817bc26171ecb3ce92e : SUCCESS in 1h 00m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcb7395ec1d34bf4aa00e428702e038b : SUCCESS in 1h 23m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a617e07a921b435fa4de414d684ecc98 : SUCCESS in 1h 24m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bae7ca8548444dd78482be01b9561bcb : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cce6325704f513124b22f41339576fc9982adae","revisions":{"8cce6325704f513124b22f41339576fc9982adae":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:13:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/718802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/718802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/718802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/718802/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:13:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:13:01.000000000","tz":0},"subject":"update constraint for XStatic-Hogan to new release 2.0.0.3","message":"update constraint for XStatic-Hogan to new release 2.0.0.3\n\nChange-Id: If139ec2d20db213308620deb233ef393ceeea367\nmeta:version: 2.0.0.3\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Commit: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Change-Id: Ib58d07f9db78a1f0c52af77397acac01fb349395\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cce6325704f513124b22f41339576fc9982adae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cce6325704f513124b22f41339576fc9982adae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
