)]}'
{"id":"openstack%2Frequirements~718803","triplet_id":"openstack%2Frequirements~master~I7088a06323dc6636fb8a23d8ad71b439c69ef41a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7088a06323dc6636fb8a23d8ad71b439c69ef41a","subject":"update constraint for python-watcherclient to new release 3.0.0","status":"MERGED","created":"2020-04-09 21:13:41.000000000","updated":"2020-04-10 08:24:56.000000000","submitted":"2020-04-10 08:24:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718803-1586507096602-1dcbafa2","meta_rev_id":"1cf105f69cb7fb75907827d7db2264bf9a81087f","_number":718803,"virtual_id_number":718803,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 08:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 23:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 02:12:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 02:12:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 23:07:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 02:12:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 08:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e623baf575fdf5c11462c92cfbeb363681a32a10","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:13:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9eef361a9d1aa3b518a77c77b71d7c046e8f4e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 22:59:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e02296b977e4a57803f8ab088f4b626 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8823dd0a752645dc8e02873560ca221a : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/156ca469a6a04c05ba0d05505fb84898 : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3898d44189e44e4289e072afed9e18bd : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13473d85bc514465ad880622be1332c6 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24c8aed03a9b4af6a65f0ccb57eb797f : SUCCESS in 8m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73b9443c315e4942a24e388801812ac3 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f5f2d90c6c246c2b374389f9e9f8fd9 : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a73210bb37ea41ee969704b209682c26 : SUCCESS in 10m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/815c73ac498b4382845d4974447b8fa1 : SUCCESS in 6m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7acf827b86514c618b48833103035b81 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49acd486deb843c2a99bdace1507969c : SUCCESS in 9m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e09550ac38f048efb66adab9e891a3e9 : SUCCESS in 15m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7c59b5391b0744b5b5ca4feef814fa4a : SUCCESS in 28m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b05881c7b7e4bd5a6788abcbb2c4312 : SUCCESS in 15m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bbb1c13bff26451f8a547547b119f805 : SUCCESS in 15m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d5468c4bf444df0b361263538ed348d : SUCCESS in 20m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8860ffdd58b64f1784571d2e526b2a66 : SUCCESS in 7m 05s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/af138e69b9874fa9a1ae4bfc5674d92f : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d5768be4624426e917070f1d9e5a0cd : SUCCESS in 10m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/32a386ba2b5741d3aab7cba63a966e58 : SUCCESS in 14m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/080a7fcdb43c45e695effadcd0507aee : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afa27e7180594ad0b3b677204f9cb021 : SUCCESS in 55m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/019b9fec45744cfb87d8cd7e8403e678 : SUCCESS in 1h 20m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4746cc96551492d9866e8388597671c : SUCCESS in 1h 17m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2cd6b003336542b59a0f8d07bdfcff88 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9a9ebde292ffce7d739ce09da8cf2fbbd4a04da9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 23:07:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"288e49134e17c546c77f4c2ed9ed1c31455a784d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 02:12:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d05471dd2269350fae650e8e32b1b51f8b57aa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 02:13:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f45120c76eef2b398b13b77ce796f0fd0ada0386","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:24:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3b9c681a403dc9d8adfb3566e1ea227f936a1562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:24:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c337a637914ea985bae125dec50b16 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db950c1fabee495c9188d2d90a5f663f : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76e718c5ab6f45889421ed20a735f324 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9026e87daaaf4b138b1c5847895d0c96 : SUCCESS in 7m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5aa94c36481405c8a70861ee571b5b1 : SUCCESS in 7m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b356c2447b04754b16030c42ac04edc : SUCCESS in 6m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0fa1a969f176447b8bea9d1859406982 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7082ccb58f234c618324c1b6a654ba27 : SUCCESS in 12m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f20461a8040441d6b1e2fb15135feaca : SUCCESS in 9m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e3054f6a5a02434b9ab3f8d8fcfc56af : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d651d98756149dc860abd549199e9b4 : SUCCESS in 10m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80c51476f8944b56a3c96da91736be82 : SUCCESS in 23m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bc861af2608847129a8b170d21c4d129 : SUCCESS in 28m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a73b5f4744c4d4ebac61f41c9ec8445 : SUCCESS in 14m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/583e6f1e58274e1688de5feda1cbb1d5 : SUCCESS in 14m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54dcd7b05e26464283e8b2e9bad2ec49 : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e73369505e88410c9a793b9b5b95e3e0 : SUCCESS in 7m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2c5e14007f83451e881feef2da78f176 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8e2d1f8d2944a1d9446b87c2fa605dc : SUCCESS in 8m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a5c304ad7b16415d9a201e080c5640aa : SUCCESS in 12m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/edef05c5f17e42adbbc58b63546955e5 : SUCCESS in 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/366326e3a74149a0b63b1cdbcbcb9b8f : SUCCESS in 52m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/52427c114039421fa4a5823d0721d5b6 : SUCCESS in 1h 12m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d29894a2c54454890db6de2cdba4850 : SUCCESS in 1h 15m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aefcb58d0e9f48d984445d6bd74ab42b : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7d446e26137e144e866dc93b1da8b88a83e9269","revisions":{"c7d446e26137e144e866dc93b1da8b88a83e9269":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:13:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/718803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/718803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/718803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/718803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/718803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/718803/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:13:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:13:40.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 3.0.0","message":"update constraint for python-watcherclient to new release 3.0.0\n\nChange-Id: I7088a06323dc6636fb8a23d8ad71b439c69ef41a\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: licanwei \u003cli.canwei2@zte.com.cn\u003e\nmeta:release:Change-Id: Iba00adcca12d0e459db72e04eef87d9fa07aacd5\nmeta:release:Code-Review+1: licanwei \u003cli.canwei2@zte.com.cn\u003e\nmeta:release:Code-Review+1: Dantali0n \u003cinfo@dantalion.nl\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7d446e26137e144e866dc93b1da8b88a83e9269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7d446e26137e144e866dc93b1da8b88a83e9269"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
