)]}'
{"id":"openstack%2Frequirements~718805","triplet_id":"openstack%2Frequirements~master~Ia937316af37aa782e2f57faa001f0b4edb66ad5a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia937316af37aa782e2f57faa001f0b4edb66ad5a","subject":"update constraint for XStatic-Spin to new release 1.2.5.3","status":"MERGED","created":"2020-04-09 21:14:24.000000000","updated":"2020-04-10 08:32:55.000000000","submitted":"2020-04-10 08:27:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718805-1586507233692-b33c4e60","meta_rev_id":"879f5495e5e72c309c169a3158f32dfb1d972b21","_number":718805,"virtual_id_number":718805,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 08:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 23:06:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 02:12:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 02:12:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 23:06:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 02:12:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 08:27:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6eb7cacf39bec74823e1c1d6ef62eff371778b95","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52e0aed9b02cb91d43276b209c110e22ed04bd6d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 23:06:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dfb8011717bd2045a289b154208c8f9339001772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 23:25:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/893b7af2f23d4d6aaa9275583a0faaf4 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d5f1d102e945e6a830b1bf81adc12b : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bda7a536f00f44928fc3bb86872a1c18 : SUCCESS in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc54b6d53e844d2581e9c1cf251376b1 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e7509a28acb4d088e7e35a06669843c : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4be725df6dd40d6ae69a502e3a32adc : SUCCESS in 6m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/048b19a179494193b430ead81418f839 : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/235122ac0e8e4940a9d7c8f6c741e3a7 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d0a5e2c18e8b467caf5116a00cf97b33 : SUCCESS in 10m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab6c3928581d423985380d5482477066 : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5cabebe027a24771bcd66f9602779c01 : SUCCESS in 13m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8cd45e621ebe415488a0573013566452 : SUCCESS in 8m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/daa0f2386fec43c5a66dcb44a139cd1e : SUCCESS in 43m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f666299891dc42f18b53ac0c892709a4 : SUCCESS in 29m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4d1ab85120444404ac7c68adc21d56d4 : SUCCESS in 14m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1404b18f2faa4cd38da506dd88bfb9fa : SUCCESS in 14m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0bc12d0e979e45fe9d9f09f6642dbda0 : SUCCESS in 19m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3d07e49841df46b986e57a4a37bb023f : SUCCESS in 6m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c50f20cf433241b2b01f34a9da547cf6 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6572e1ede2b84effbfd97b4af1f03994 : SUCCESS in 9m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e6f43fb597b14453816c77e50265c93e : SUCCESS in 11m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3db5ec37826840b28de90d4b5d39c149 : SUCCESS in 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98928b8639f94f37bb2de645bd4a23a4 : SUCCESS in 48m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf29c5ed615949f1822ede9dbc97bd31 : SUCCESS in 1h 16m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7375239bee074d0d833220f42be174ea : SUCCESS in 1h 24m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cbd509b1e11b4f9e8b15866511e99842 : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3f4d2a786481c666e872d62422ece687a9b3c6c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 02:12:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"75cda535a496e6874ca05dd5bcefb7f3eacf5388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 02:15:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d215a79745d88f95b6423b9cbccfedd2d9701c82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:27:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ab4733458e4f0cba5d63916a590ff4 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0416ea61d0cc41ca8d8d094e81993bd2 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0c04660148c4782b023dc1bca84a08a : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5954a7bdf64b4bca99a7e650412bcde1 : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea54f1fb79154cab8d8c56f75dcc61d1 : SUCCESS in 6m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2ffa80e06004195a862a3e5562c8cf0 : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e50f7bf6c78d4a3b83884c222000a02c : SUCCESS in 16m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cff501725d6e4deeb1b0f805023a2d13 : SUCCESS in 13m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e905f378a5c443dda0b0a1a3423923a7 : SUCCESS in 7m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ec83b927f3994e5aa4f0b41fc8a9e794 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e8b2f496ffb44d8ad180888ed99ef2e : SUCCESS in 11m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9e644fab93dc439ab8420eb051f8279f : SUCCESS in 17m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f9bc93edd17420aa5b1be8a81166d39 : SUCCESS in 30m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fc5aa203365748e59cecad23f571230a : SUCCESS in 17m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/69fde40386fa4db19889079eecefd60c : SUCCESS in 16m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5adefe26cf234063918d999d491e9cb7 : SUCCESS in 20m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ff644d5420f64d44903cd4156d3923b9 : SUCCESS in 6m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/563a66976a49486da995bcac6befa540 : SUCCESS in 7m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f0a6378a9c34bbaac75927899722f16 : SUCCESS in 10m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b76cbb20a7274ffd9521458cb8a149e2 : SUCCESS in 19m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c425271e13bf438e94262f353cef7aac : SUCCESS in 19m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9210863b383d44728eb0842b9ffecef8 : SUCCESS in 53m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/21b9e98c51fe406ca3c090894fd509ab : SUCCESS in 1h 20m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/737652e8c6aa47dc90e14eb1dfb2e983 : SUCCESS in 1h 27m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ef63d6acec1f47b6bc8c859918660230 : SUCCESS in 1h 20m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b3ed243e9fb473b2cc4cfbd18c121bb7f8d70611","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:27:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3508a07cb552464accbfe67c9888ab7721be266d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:32:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3cfc63ce4034194b6328c5f9068eb01 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17f3c4c05a92f9d0eabff9f8edbb5f558ecbf45b","revisions":{"17f3c4c05a92f9d0eabff9f8edbb5f558ecbf45b":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:14:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/718805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/718805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/718805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/718805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/718805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/718805/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:14:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:14:21.000000000","tz":0},"subject":"update constraint for XStatic-Spin to new release 1.2.5.3","message":"update constraint for XStatic-Spin to new release 1.2.5.3\n\nChange-Id: Ia937316af37aa782e2f57faa001f0b4edb66ad5a\nmeta:version: 1.2.5.3\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Commit: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Change-Id: Ib58d07f9db78a1f0c52af77397acac01fb349395\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17f3c4c05a92f9d0eabff9f8edbb5f558ecbf45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17f3c4c05a92f9d0eabff9f8edbb5f558ecbf45b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
