)]}'
{"id":"openstack%2Frequirements~718807","triplet_id":"openstack%2Frequirements~master~I568747cbbe23b832560b45c8b3362cbdb9b908b4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I568747cbbe23b832560b45c8b3362cbdb9b908b4","subject":"update constraint for XStatic-Bootstrap-Datepicker to new release 1.4.0.0","status":"MERGED","created":"2020-04-09 21:15:14.000000000","updated":"2020-04-10 08:38:42.000000000","submitted":"2020-04-10 08:35:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718807-1586507745021-9e742fff","meta_rev_id":"cacf3b012e13cf46b2eb7b7be980590f604584c5","_number":718807,"virtual_id_number":718807,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 08:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 23:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 02:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 02:13:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 23:06:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 02:13:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 08:35:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc9b9af4c5a0549904f1cefa0d4c44a22b82ef89","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:15:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a87046a5de639b6056ddd1c2021276a724062487","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 23:06:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"91156cb74867966053d9d260d9e6350b155f26bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 23:24:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a30e4be7dc4124be02381048d4b266 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6202aa742cdc42a696071684770ff320 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2d3e212c77046cd8c34d3b33d1dd4c1 : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6611e369998408f931c33e93640041d : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2ac06d437c247369200edca741f3ec6 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/146bbbda66aa4a63ade64b0ec82cf898 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/221d716164f8452baf1b38db86ff2e80 : SUCCESS in 3m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ae862f28e3314b48be252f908703679f : SUCCESS in 7m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/16d22c09fdfa4abc822368583db949bb : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cd34472b67db43b6baa07d324cac20dc : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bdf620df9d914e869d80c8eac3cc1a68 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/865d0744b65a4332b00b3133d76f1667 : SUCCESS in 8m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/37c45365a3a74a8f83cea10a939dbfef : SUCCESS in 23m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9bb6bb186a2d4b88a5140cbe95a5fd5a : SUCCESS in 35m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c634f4f2e9b42ecb63e5eaa9de155a6 : SUCCESS in 18m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/56292149480c44c0951c3b7d8952b521 : SUCCESS in 13m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08aa0e36d7b042b8bac45c790ec775c3 : SUCCESS in 19m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/62ec71a877104eb09965ba8f289a41df : SUCCESS in 6m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cf51042995d24af08b48cec9b69ee99a : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8d14e9f86b24e4c8325ca9d58c350f5 : SUCCESS in 8m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8d69b34eb272447e900b530f80ed8e13 : SUCCESS in 13m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/72061bfdd14e494fb22db5924ccf5764 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c6271072d4046059c5377380d81a74e : SUCCESS in 54m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/141d517249b043d48e505a5d688bed6f : SUCCESS in 1h 16m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38ac89e9f5474fb3a34af3b482acba3a : SUCCESS in 1h 18m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/30972626f94a40f68cf2afad3ed828ed : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7a0ba96131c9c44494dd0f69c46ef8a525be90f0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 02:13:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f744934033cf88adeba8ca1813ad3a3a3285c4fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 02:15:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d75b397248add7084829990fb9f862eabd5abd0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:35:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76153b0944884a8a9a45f3837afa12e5 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45bc399356524d72b658d93cf227e397 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4af16e7c20ba4cdd94a57acaf7c84b5a : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91c317c880114586b0ee5079b211fed7 : SUCCESS in 8m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6a88b1d94a540a28e7c553657013016 : SUCCESS in 7m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e717ee40e94b4e5ab22055f808f32cc8 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d62dc8f83c3b4b31a37f8fdf959e93a0 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/348ab7478bc5473592c612d90b44571f : SUCCESS in 13m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6bb36bc9401f401d9886426c2662fe7f : SUCCESS in 8m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aa37706d1bbd4cd585a4effe44515260 : SUCCESS in 14m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fee06748ce0443fb1677cc7db82dc80 : SUCCESS in 10m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b88038726d6427b96a08604c537a3d3 : SUCCESS in 15m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f760980a3304e54a0a20c4a49341cb8 : SUCCESS in 27m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4c1c4a793d3142278dadf9499bec6f94 : SUCCESS in 15m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/97f3f19df527411ca348f2e56150ecbb : SUCCESS in 15m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a6873d813814b929a33f4e4b4c8202f : SUCCESS in 21m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/22fb7f144ffa4afdbb4fb44eecccc523 : SUCCESS in 9m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/29586cf4f78c4b4a9a4a01fbb3d15b72 : SUCCESS in 11m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8dc8cfacd99c41ffbc932a1a5681d5ec : SUCCESS in 10m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/137bb0ddc0ee487cb41ca5f95fac123f : SUCCESS in 14m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0afe31b2fdb74fcf8b00f93b964c125d : SUCCESS in 16m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25ca5168f9344c709451f5f8dedc360d : SUCCESS in 1h 05m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a01712e99da7483ca7f6050abb45ac69 : SUCCESS in 1h 20m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/851632bfcf574957b0f57bf551baad4a : SUCCESS in 1h 17m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ac41b549685845298d5c4df74fc1c258 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9ba10f9d2e73235ebf7b4365384d4a43229d2966","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:35:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ea4f70f3c7ce365eb554afdec9c276d4e881008f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:38:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44a2a09d5f548adabc346c44cf87f6d : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c87317831c53891a2b4492d251d8ccc6dee3286","revisions":{"2c87317831c53891a2b4492d251d8ccc6dee3286":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:15:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/718807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/718807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/718807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/718807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/718807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/718807/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:15:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:15:11.000000000","tz":0},"subject":"update constraint for XStatic-Bootstrap-Datepicker to new release 1.4.0.0","message":"update constraint for XStatic-Bootstrap-Datepicker to new release 1.4.0.0\n\nChange-Id: I568747cbbe23b832560b45c8b3362cbdb9b908b4\nmeta:version: 1.4.0.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Commit: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Change-Id: Ib58d07f9db78a1f0c52af77397acac01fb349395\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c87317831c53891a2b4492d251d8ccc6dee3286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c87317831c53891a2b4492d251d8ccc6dee3286"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
