)]}'
{"id":"openstack%2Frequirements~718808","triplet_id":"openstack%2Frequirements~master~I6b18cc86f9994a3ee0510435a406fd3df9c6e2b4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6b18cc86f9994a3ee0510435a406fd3df9c6e2b4","subject":"update constraint for XStatic-JQuery-Migrate to new release 1.2.1.2","status":"MERGED","created":"2020-04-09 21:15:44.000000000","updated":"2020-04-10 08:37:24.000000000","submitted":"2020-04-10 08:35:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718808-1586507743950-1fee8ca4","meta_rev_id":"f26450398d96492a3457ffa925f214aaa632af0c","_number":718808,"virtual_id_number":718808,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 08:35:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 23:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 02:13:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 02:13:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 23:06:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 02:13:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 08:35:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad700d03d6971eab730f1c47f1da3b27a61562ab","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:15:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e958200e495f384e70439356038ef993fc838c67","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 23:06:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ace448e332208551ad4e84e6c6651d8e4251fffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 23:26:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333a6fcbffc54d32b5acc23431039d55 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/031473d20dc14c57b751c013ca6d8dd5 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17e428b0e10c496eac0853bbd5518882 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d892090bb9e45cc900a895fb5c7dca7 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a12d4d68ee72479e8d9e61d6dbf7464a : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de6a843a643e4b91807e105fc6053d4b : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80b7f1c4eb6749c8a54676b3df6daa52 : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/73f31cf3271140598ff1f3557ebb78c1 : SUCCESS in 11m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5324ddd82c0240fc9b48b45dd14723bd : SUCCESS in 10m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7697e6d29beb4fd0a33ab77d600fde34 : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/52ddfd20a6e446dcab288c8938184666 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bc92c999d8f47f19bf434c17060cb45 : SUCCESS in 9m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3c8e989066546b88985f6c802dca319 : SUCCESS in 14m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/77ff1b981a694a80a05244447f05a735 : SUCCESS in 29m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/98c000b50a654a4ab4d8d5566860b2dc : SUCCESS in 14m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7617743e7e81432aa5033baf33cbe6e9 : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/baf4940579e041c7b604e738ed2e1c5c : SUCCESS in 17m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a393b112ec604b5e86aa04d600407413 : SUCCESS in 5m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fab961a4e047448480b549ac22c9b56b : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5d3239e7dae494a8be7eb57ad8f8058 : SUCCESS in 7m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dcf391fc93c0485abd77d12719eb57f5 : SUCCESS in 11m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/07b8f3b8341743b4a17ac53e513044ce : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5228d2a7685741a5ae267045d3bd5c01 : SUCCESS in 51m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/84d8b137e3e9483f9d040c3970bbc779 : SUCCESS in 1h 19m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94203c584d494caeb8ee3f66c2caf1be : SUCCESS in 1h 16m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/84e392ca4cce4d03adec14111f4756c9 : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":1},{"id":"91666ff17b4d350c002fa8d0246cfe032cdbde23","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 02:13:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3aa1490841c05876867147e0848d98fdc3cceef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 02:15:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16db0c5ee9c393a28d33d1500a793e74aff943dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:35:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3baef80c43af4074947a1738829f6ba7 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c9f0efd51c04f2b8178c02575dab811 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ced82188d1724df48a3eca4ee09ee2d3 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd05399bbfa54ee1a9fe78cba7ff42bd : SUCCESS in 8m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/196f57d17eae4145a4385f4daab680a0 : SUCCESS in 6m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0291d2bd81574915958fe7e90a34d527 : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0e4bf3b79a644f4eb2cfe1c5987c5d78 : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a43bc5b79a24983ba7e37f01e0cfa53 : SUCCESS in 12m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/207994617bfc4fe69d2bca24b44d1a08 : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a7a04459eef345d9a2515d010e383213 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23b7972d20d1469b9fbc3e4da292d3bf : SUCCESS in 14m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/92954cd167f844dba62452e3878842ad : SUCCESS in 26m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/957453934f654dd49637b67bb0bb78f3 : SUCCESS in 29m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/824818762852480ab2514e7c8c9d2dc5 : SUCCESS in 15m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/99ebccec8af14e27aa15937e1d07e51b : SUCCESS in 15m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31220d404b264c628785bc677a629db1 : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2cfbdc21062f4b8e861d20d6b35ab081 : SUCCESS in 7m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c5f86256ddef45b9be9a2f96ebe6b0cf : SUCCESS in 9m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5bd20ad96f474e5da2ace83388a9f2db : SUCCESS in 9m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0adae0fba6f34539b722c50f53f90fc6 : SUCCESS in 12m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/577238409fc448918bb6b05cd05d26ae : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a54f2c14496a4a4cbfa19c8a8cff3e24 : SUCCESS in 57m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f51afc17f1e240afb9c83d10e77d5fe4 : SUCCESS in 1h 18m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c3d38b5555142499a86037088b0c7ad : SUCCESS in 1h 31m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/433170b06f594af59940b6678a4870fe : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":1},{"id":"194c7f5605ec99be9f0d0c437a85f7ce3a1fe113","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:35:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dfc23a02ade2a7d2457ce5d75305c90c1558a2ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:37:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22563b9e5c2240a4b490118b1d70f70a : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a2f0106b1ff73fc157d449ddef59cf32387b230","revisions":{"6a2f0106b1ff73fc157d449ddef59cf32387b230":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:15:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/718808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/718808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/718808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/718808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/718808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/718808/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:15:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:15:42.000000000","tz":0},"subject":"update constraint for XStatic-JQuery-Migrate to new release 1.2.1.2","message":"update constraint for XStatic-JQuery-Migrate to new release 1.2.1.2\n\nChange-Id: I6b18cc86f9994a3ee0510435a406fd3df9c6e2b4\nmeta:version: 1.2.1.2\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Commit: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Change-Id: Ib58d07f9db78a1f0c52af77397acac01fb349395\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a2f0106b1ff73fc157d449ddef59cf32387b230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a2f0106b1ff73fc157d449ddef59cf32387b230"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
