)]}'
{"id":"openstack%2Frequirements~718809","triplet_id":"openstack%2Frequirements~master~Ic10d49d72f7c60b44006caf5574c4c5cf22a6aa7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic10d49d72f7c60b44006caf5574c4c5cf22a6aa7","subject":"update constraint for tripleo-common to new release 12.3.0","status":"MERGED","created":"2020-04-09 21:16:12.000000000","updated":"2020-04-10 09:45:47.000000000","submitted":"2020-04-10 09:43:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718809-1586511786017-1fee8ca4","meta_rev_id":"ae270bfedf2257e35ada4ca9c4a7792739ac9991","_number":718809,"virtual_id_number":718809,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 09:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 23:06:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 05:54:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 05:54:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 23:06:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 05:54:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 09:43:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60c871dd46233c7e4caeed42687a4653e2698bcc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-09 21:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"158ebf1cec3410dcc642151458f700b2e940549a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 23:06:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc3c96ccc1382858638c0a027f07bbe60552621e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 23:42:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6021d5fd43247c4bcc22576031db0a2 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1dd3e2d4104b00acdb4f1e0400c1e4 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22ae18cdf6744f459fa070644ac21860 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/151b9a54916d4c9eb2ff5cfc100f479c : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b38a00b5606b4be6a49630b9654e2c83 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab3afec72b3d45688c048284c3a75d5c : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1475c59eef5340579ab75fbbdc87f318 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/57ae839460804b819d7d29cad3694414 : SUCCESS in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/780ff401b98743cdbd57c27fc2557273 : SUCCESS in 10m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/833481eb17d2458bb390ef7b80783681 : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5d8f29eca9b34fc9a23a31b742471c83 : SUCCESS in 12m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/415227a5759f456193552a3331c7d06a : SUCCESS in 8m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9ad194b245c046ab8d7f6f05e27fcfef : SUCCESS in 13m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/324e86dc671d45c0ad1661079f9475b6 : SUCCESS in 30m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2fc53952a64a4f399d9e96c97f65f887 : SUCCESS in 14m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cd74919917b14851a134ab39497dc9e3 : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/378e6d39260842e39324e53acea76fe4 : SUCCESS in 18m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ed475c345894490aabfe1391d4d7e94c : SUCCESS in 6m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c5bdac6b2eb942d3bcc18ee313129e1a : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76346c2559564861937fc2beb4c5036e : SUCCESS in 8m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6885082d789a443b93e27ba632aaad04 : SUCCESS in 15m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e73a1aa2468c4aa6a951fcb0059997eb : SUCCESS in 16m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c2e4e3632664532bc0afe607f876d5d : SUCCESS in 54m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29538e0feb8f41fd97305f1d0b182699 : FAILURE in 1h 03m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0532d0377c9454fa6a38964e012cc1d : SUCCESS in 1h 35m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e0bb9df992e44558a6e3f3f1e281fdf0 : SUCCESS in 1h 22m 44s","accounts_in_message":[],"_revision_number":1},{"id":"39a1c541d869fcc416ad713ac5bb85824b67bd32","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 02:13:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a47a50accb5252bcb9d91aa42cacf65c4dbf5ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 04:11:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6372b3446342e7a343f3eea1e8800a : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1617b7f8e02b49b2a5d841a04b589b4b : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62be1ebe2f9747908ad41f52859f624d : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54d9fbbc76964f2098be7e0a47613a75 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79d11d75c5bb4c6e9f7f599f2c57023f : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05bf69a8443f48aba7ac8980f130051a : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8edf1d4b47304df38ce3da034be31728 : SUCCESS in 2m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/056ed4b3892d4bf1b28c35877257316d : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c354dc23bba04e6eb3d9f927ef59495e : SUCCESS in 10m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/342e18a0a93d45bb85b9c06afd3b91c4 : SUCCESS in 6m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ce9ff592cb4741a48732e13bffe9b57d : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ebfa37bfd7a4d1fa1f3f46292e7e5a1 : SUCCESS in 7m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ee1f999cfee42c08ee7fdc85aa8fd68 : SUCCESS in 15m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e19f757100f04cbbb2901a4cbc8de676 : SUCCESS in 24m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3c0d146030f94f2da27b5822fe524ab4 : SUCCESS in 11m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/453280004ef249f69d77ba262ac9caee : SUCCESS in 14m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b21dc70ad2f54c2ab2244dfcfcb1af86 : SUCCESS in 15m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/82e5cf8e921f4160867bed2f5fc9e906 : SUCCESS in 4m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/418b942de37c483c8c3d3e09fac93ac4 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4a42c0a16714d59a2f383aafcd0d910 : SUCCESS in 9m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a8abf601548b4ba0b963335f58c6f038 : SUCCESS in 10m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3b0c47986cd14cd99afb6e0cc1786805 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e729a8fb07114645a373bbed020d68a5 : SUCCESS in 52m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/680d4219a99d45f3b13e4374f0339b78 : SUCCESS in 1h 21m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0643adcec8844696b617931072be0b39 : SUCCESS in 1h 26m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/694169f45d3c41538dc1817d7aea4b10 : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"42399b363a6440cc56b4d0d7f119e47b3020c56b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 05:54:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"196b5cf1ef3908940ffd18323584fe9275fbce00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 07:18:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6a77b061399a219b6174d7587ac6836528ff87a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 09:43:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85de0e0e3e114be79d48731e4f6fe799 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9574b73114204cb18a2a7b36cf3f8142 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37545f503e434d8990ae9162c9ff44b2 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12041d0816a34d1fa8fcc2ce693c3221 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e21539bebec243e8a8a1555d1cf05081 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/528674cc754c40ea91d26396cd905d81 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5e8275ff8af64d58a1d643d75ecad653 : SUCCESS in 16m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/650b96f62cbc4e39b7dc174aa9b0d473 : SUCCESS in 17m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/322195573dd640d18ad7905157308ef3 : SUCCESS in 7m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad65762ee8ee418289048467f4bde7d3 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/959fa83a4f404f6e9df9df424a37714d : SUCCESS in 14m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f0715d001c44147808f562e5a229d62 : SUCCESS in 30m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/acb90afd565e4f378a32ce10d211bec4 : SUCCESS in 32m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e0e767fdaacd4d01985e9ed10db14a8c : SUCCESS in 19m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ee284b5bd7ca4024b7699c7135ac1f96 : SUCCESS in 14m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93b786321d924606a579e1963820dffb : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dd3ebb4e65ed430caacb0996f0101cce : SUCCESS in 4m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0dd54d35596245bfb7620d90b31c8cd8 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/557999a29ceb4f9384f5c744cdc983ab : SUCCESS in 7m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9e0b9c6b8594440581925e43415d7687 : SUCCESS in 13m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a6ef5d3fbcbf420c8bade19571572c24 : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/362aeac5caff47ff892d36aa5ab221cf : SUCCESS in 51m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c86891673eaa412281d8718d6a41d19a : SUCCESS in 1h 12m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75c1828c359c484982e87f34767ed6a3 : SUCCESS in 1h 21m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e171d16852174b9b91d7ee1cac701075 : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e92d7312bb2c1dbe5d726aa8bef2733392c2e68a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 09:43:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aef2da836495193e5fe28509fa7f6d21da96c8f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 09:45:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558932890e3c42559ad554b9a5e2c03e : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da486a77e92201ffe8e3a556e6784a23f655e457","revisions":{"da486a77e92201ffe8e3a556e6784a23f655e457":{"kind":"REWORK","_number":1,"created":"2020-04-09 21:16:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/718809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/718809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/718809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/718809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/718809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/718809/1"}}},"commit":{"parents":[{"commit":"de3d3ac1779b849c010ce66ba7a2af6dfe5f862a","subject":"Merge \"update constraint for python-mistralclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3d3ac1779b849c010ce66ba7a2af6dfe5f862a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:16:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-09 21:16:07.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 12.3.0","message":"update constraint for tripleo-common to new release 12.3.0\n\nChange-Id: Ic10d49d72f7c60b44006caf5574c4c5cf22a6aa7\nmeta:version: 12.3.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Change-Id: Ia52230140b5124423ca904fdc1e46f67b6e569bc\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da486a77e92201ffe8e3a556e6784a23f655e457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da486a77e92201ffe8e3a556e6784a23f655e457"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
