)]}'
{"id":"openstack%2Frequirements~718993","triplet_id":"openstack%2Frequirements~master~Ibcf3f8acd490e49a98bd760d580c2aaf2f3763f8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibcf3f8acd490e49a98bd760d580c2aaf2f3763f8","subject":"update constraint for python-masakariclient to new release 6.0.0","status":"MERGED","created":"2020-04-10 12:33:50.000000000","updated":"2020-04-10 18:12:36.000000000","submitted":"2020-04-10 18:11:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718993-1586542273022-eaa21e4c","meta_rev_id":"c1181189562cf39eadce51a4c790a42ac5a616f7","_number":718993,"virtual_id_number":718993,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 18:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-10 13:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 16:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 16:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 13:49:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 16:49:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 18:11:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"583a3b8df70723cdbf0ba06c7894e4854f2369f2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-10 12:33:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d052d45337775b8050d040c46f27f23b87c622bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 13:49:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"580bc6df389c7c4c35fb82d7a5c4efe6a63e2215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 14:01:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f6270bb063a44c3a5356d8da2c432d2 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6c09026b62405c9bc8e71c9da500dd : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24daf490257541418a4726fd13146af2 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc363ce5995d4826b22c5dac6c318469 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fa575c7202e4a8bae6177f5b2a1e852 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad3ce26c99cc4549b6c6851b5b5965e5 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52605414732b4f708dd98a1fd0343d64 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c786e629d11646d69c8fa43193e4b6ac : SUCCESS in 7m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ef460aa64ade4acc9ce8c7e67aa47eaf : SUCCESS in 8m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbf9501677974865a52a1b301848574e : SUCCESS in 5m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a23074dfe09849918a4e24a2ae5443c8 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21e61a9905d6432da2b5f5fa3e575235 : SUCCESS in 8m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/84a2c7ce697246198ecec29236f757a4 : SUCCESS in 22m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1f6a5d8ccad74316a0de061ec092c766 : SUCCESS in 27m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/72076d6cb96c4fef907d755a4d0b259e : SUCCESS in 15m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5e2121d093b04919929fc4b8a8418fd2 : SUCCESS in 13m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95306a9bf5e34bc0a97a5e6c80b64dad : SUCCESS in 21m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/565c9d8578a8489d957f5d462b8ff1a9 : SUCCESS in 4m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3f63d05a3c6c4604af132ad7cbaa2c17 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e516f804f3b2480bb0a46cbaedeab8ba : SUCCESS in 6m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9c42034341bb4b039b2c7defd0942d74 : SUCCESS in 14m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bc1d0440830c40df95cb282eb0cc18bb : SUCCESS in 14m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47418b2c66a147f6b67028d564abef6d : SUCCESS in 50m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b453771900f4cad8ea9cc89330764eb : SUCCESS in 1h 16m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85168eddf5194323a4463e7f2ff6192b : SUCCESS in 1h 26m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1b83a09231d244969fcd597585202a98 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cdc404e33dec0ac2670906af764d4beb92d7e4d2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 16:49:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10a6fa305c82eea2916ddf582f7882e6faef629e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 16:49:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"289a851cd7db835f03bb09598aacb45254f4bcfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 18:11:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65fba815582e48b4927b7020d2b03df2 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b6a482e2f94147aabf4b13d89d8c86 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76e6ade231854d55997c37766655ffb2 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de91b686c300479aa24a3bd4d9b7e2d5 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efa896d0370c40e298070a442a88cde0 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/334d0132c9924a6e9b91a0f090b4064e : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37f6b90b170a443588e94d15a0729f4b : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a5e209747d341eba9285a6721006083 : SUCCESS in 10m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9d8e9409a4fa485599d50f0857f78190 : SUCCESS in 7m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92371658680946c5aeac4ac833c0a1ce : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04b50c2ab38548a4bd56f1534b811cc1 : SUCCESS in 11m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/29141c559a1947aca5f1b0603763941d : SUCCESS in 36m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c430fbf360ba41a284aee22c726460e9 : SUCCESS in 32m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a42954392c44410aa54c496c2120d5af : SUCCESS in 22m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d3393a86400e4cc39f229a1589db9e6b : SUCCESS in 22m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfe09885e1e940b4bed66c9db9042ad2 : SUCCESS in 25m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2d783501edf14712b8b82cf29ac688f6 : SUCCESS in 5m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/81cf08333fe34092a7cd4dba73308eb8 : SUCCESS in 6m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/271a62fb100e46f2b7c8994b79ca5204 : SUCCESS in 7m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/85e7ab7a24ef43f1a6bcfbb525f8aad7 : SUCCESS in 11m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21587cdb55524336aa23c79424f72b14 : SUCCESS in 22m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f29e31e1df984b77beb30f687336e373 : SUCCESS in 52m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/304b203aeeab4143a95dae56c83f13a2 : SUCCESS in 1h 11m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/904255757cee416e97909454cbaece5d : SUCCESS in 1h 20m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ce2764c83afe4450a350ca91b2929721 : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9c77d2ec4374bfbc67f3d28d2075326bde675ca9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 18:11:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"142fe2e8b7240090bc546fadc675ec7ebd9cf00f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 18:12:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f619812e6c45405e8d069e9678253a30 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fd30665b10555af6530ea1b7762c8a5e00e5e31","revisions":{"0fd30665b10555af6530ea1b7762c8a5e00e5e31":{"kind":"REWORK","_number":1,"created":"2020-04-10 12:33:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/718993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/718993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/718993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/718993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/718993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/718993/1"}}},"commit":{"parents":[{"commit":"bc51793ea79968e1fa42088ee6ce08ce4a56ecb2","subject":"Merge \"update constraint for tripleo-common to new release 12.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc51793ea79968e1fa42088ee6ce08ce4a56ecb2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-10 12:33:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-10 12:33:49.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 6.0.0","message":"update constraint for python-masakariclient to new release 6.0.0\n\nChange-Id: Ibcf3f8acd490e49a98bd760d580c2aaf2f3763f8\nmeta:version: 6.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sampaht Priyankara \u003csam47priya@gmail.com\u003e\nmeta:release:Change-Id: I22ab1585326822d5ff76de474e4cbb00360243b6\nmeta:release:Code-Review+1: Sampath Priyankara (samP) \u003csam47priya@gmail.com\u003e\nmeta:release:Code-Review+1: Tushar Patil \u003ctushar.vitthal.patil@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fd30665b10555af6530ea1b7762c8a5e00e5e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fd30665b10555af6530ea1b7762c8a5e00e5e31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
