)]}'
{"id":"openstack%2Frequirements~718996","triplet_id":"openstack%2Frequirements~master~I52738cbf8ac7597c33a4a06c9a60f59eabd57801","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I52738cbf8ac7597c33a4a06c9a60f59eabd57801","subject":"update constraint for python-tackerclient to new release 1.1.0","status":"MERGED","created":"2020-04-10 12:38:03.000000000","updated":"2020-04-10 18:17:44.000000000","submitted":"2020-04-10 18:16:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718996-1586542579028-7a5aba1f","meta_rev_id":"ebc9be35dc09e4302f4bf4121e520ebabb5b1ef8","_number":718996,"virtual_id_number":718996,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 18:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-10 13:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 16:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 16:49:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 13:49:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 16:49:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-10 18:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa344e410c4806a4cf181fdf402b631917ef77f9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-10 12:38:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"685d7e4dc4af96e0bc12ce91b41efe7b428a2430","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 13:49:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"baf09b85f2ad6d8c6e442857148b360fd311ad68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 14:10:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610b8fceaa1240cd800726d4f0e1e21b : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a970bd89c774f44872189edd01ce790 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5298f09f59da4bfb9eb54afc74ac40cc : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78d2162a320f4ef9b74a460d2ce00515 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fde3584c29a4401dab75411111a6cf9c : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/633502822fc2481399333a6a77a102c8 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0ea28c8f2a44973a90fbce53e096023 : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/525bf1c705f343508c57579ac6ec1f7a : SUCCESS in 7m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/64e0e2c5c6ea409a86cc1a47d26eafc2 : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0774571cc5a4a84baad479e01ca0dbf : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9a9221a6565848aab7eb58e78d39baae : SUCCESS in 8m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a262fc035944ed38a1de8addf689a36 : SUCCESS in 7m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d897c9c9868045988a77c20d5a3b8da9 : SUCCESS in 18m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d1b4fd7b0f314d8390bd7a0415e218b1 : SUCCESS in 20m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb0b6ab58a734faab0a898e6612c22ed : SUCCESS in 15m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/13e380d863a745cfb6747533fd9f4a42 : SUCCESS in 12m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2cbbcb0141ed4237a7b1612287b0f5df : SUCCESS in 14m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0452aeb74cb84bb3b99da1509568bb9d : SUCCESS in 4m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1638169e14684bfe8db932b5a16a2b2b : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52f6ff14ed4e4790adc09b1ef5ad464e : SUCCESS in 6m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d9037bdbba1f48b8b7b5c213aa338bae : SUCCESS in 11m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e9980d1eb9c4ce5a4382d0b63f63a90 : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60a67bbaa50344aa98bce2d2733159c8 : SUCCESS in 51m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/37e400d68efd45b9b3c8e56f97727d5b : SUCCESS in 1h 15m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47c62b14cef14d668e889e4bc81b50b2 : SUCCESS in 1h 30m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c42e272012b6437fa8c9ed07d340aed5 : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c4b1a7702dbb91122a1306dc5c8f13e55076691f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-10 16:49:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7c10973658b04fddce97b11de71010495af7a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 16:49:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1db306cda8107c51b08b14eb064d6d312c9304b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 18:16:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a16fb89d33e249a489c048937f70762c : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3821702101234159bad65c9d2d1a3a43 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/838db98716184f68ab7caea45d9e73cd : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ae30f2ea5904032a90125358573d993 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c01564e4d8b74b68a6739e5bcf5d840b : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/141f77caff0d47fe84456db1123f406c : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/658849445eec4b4e8b767d1db4e94053 : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3bc2cbbb93cb40349d106eee14ecd3ff : SUCCESS in 10m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5cfac89ed6f5495c9987a86c17c610ee : SUCCESS in 5m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5720905bd4504a9295b652283f4f2663 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f743e069f0a644b0a6aedeff8f5523ab : SUCCESS in 7m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4ff7469252a445babf9912c280968ef2 : SUCCESS in 15m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/238d9a4bc9ff44a0a656b9c269c80203 : SUCCESS in 31m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f5df8781c1e84776a689b28c4e71bc38 : SUCCESS in 11m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ec1521a74fce475586c1cf9f17e2adc3 : SUCCESS in 15m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcca0688411e46fe97520e2faff25992 : SUCCESS in 19m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b817ab7cf9cd4655b281d4506c61060a : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e678e7cdcc3d4a6b8ca8063b0cf7bcef : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eed75f748dbb41f5912224e84d076636 : SUCCESS in 7m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/84bcfa70d8a54d06aceef83e57a371b8 : SUCCESS in 11m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec0a3d8f6f1947a8b4fe316c45075807 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3753fd9019143c1a8335bff52bbe9f2 : SUCCESS in 56m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a93bbef76a142cda8e8b7e712ed13d9 : SUCCESS in 1h 18m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d03833988fcc4783b6a097cab723df9a : SUCCESS in 1h 19m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f5eb87b463a740729d6b116f3ad492da : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"001566e78f49b8dbb7aaa9bfc3fe8679955ee9e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 18:16:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"677a8f8ab87b5471cbcdf011ae8c88f7f2521ae1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 18:17:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab75ae03ab149359b93f29429513273 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa98f94ee67fbf0ea290a03bf0fdbbc8050ca3f3","revisions":{"fa98f94ee67fbf0ea290a03bf0fdbbc8050ca3f3":{"kind":"REWORK","_number":1,"created":"2020-04-10 12:38:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/718996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/718996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/718996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/718996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/718996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/718996/1"}}},"commit":{"parents":[{"commit":"bc51793ea79968e1fa42088ee6ce08ce4a56ecb2","subject":"Merge \"update constraint for tripleo-common to new release 12.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc51793ea79968e1fa42088ee6ce08ce4a56ecb2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-10 12:38:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-10 12:38:02.000000000","tz":0},"subject":"update constraint for python-tackerclient to new release 1.1.0","message":"update constraint for python-tackerclient to new release 1.1.0\n\nChange-Id: I52738cbf8ac7597c33a4a06c9a60f59eabd57801\nmeta:version: 1.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: dharmendra kushwaha \u003cdharmendra.kushwaha@gmail.com\u003e\nmeta:release:Change-Id: Ibfc9469683fa9019d21c42b081681833193602cf\nmeta:release:Co-Authored-By: Dharmendra Kushwaha \u003cdharmendra.kushwaha@gmail.com\u003e\nmeta:release:Code-Review+1: dharmendra kushwaha \u003cdharmendra.kushwaha@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa98f94ee67fbf0ea290a03bf0fdbbc8050ca3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa98f94ee67fbf0ea290a03bf0fdbbc8050ca3f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
