)]}'
{"id":"openstack%2Frequirements~719306","triplet_id":"openstack%2Frequirements~master~I5ef713d22bbe570ae9c78ac00dba5fe0b370c17e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5ef713d22bbe570ae9c78ac00dba5fe0b370c17e","subject":"update constraint for python-zunclient to new release 4.0.1","status":"MERGED","created":"2020-04-11 19:09:27.000000000","updated":"2020-04-12 09:58:55.000000000","submitted":"2020-04-12 09:57:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719306-1586685458292-11ac1cc1","meta_rev_id":"e48e17f057d4f97646ea81d1cfa87e88b39542a8","_number":719306,"virtual_id_number":719306,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-12 09:57:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-12 00:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-12 00:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-11 19:13:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-12 00:10:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-12 09:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52f7ff7033a9bad3b0231e5d6c0700ce17f56b68","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-11 19:09:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d7acbf8f733b60e2aad5cc1fe9424176d48205e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-11 19:13:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2128b0a391da5ab04a1e30e1ec94acf3a9ecca5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 20:35:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb2afe2f1394798ae1b11a3526cedcd : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc84cf2919924886b0447a36d03a15df : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc8cfc3577854bfd938954ede237e70d : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1390cac8c1a745a7a2a6fe1fcb50e963 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa4c6f3b2d9347b5accd4634e0fe286e : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/410b680679e9497c82c8127480287437 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/487704a3f5464592bfb881f4e8121933 : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cf1ce5dfd16740d98b6426a27e324124 : SUCCESS in 11m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/292f8cfdbc18416bab1c478ca7c61cf4 : SUCCESS in 17m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fab9b71ddb9d43f0ae1431ff7b9dc024 : SUCCESS in 7m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5a92c96838244917bfc4310db13c9f55 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df73b8b1601e412f84b9d744381330e2 : SUCCESS in 7m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f9644f28ae164bcd960405dc0b6c264d : SUCCESS in 35m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c8bef6d227694fc2a869354604268c83 : SUCCESS in 34m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d46031ee45e0492082c96654990a8278 : SUCCESS in 23m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2a8f8e212a96408f8a21994a415bbd12 : SUCCESS in 24m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a68a1e60164a4a6c95e8775517f531f9 : SUCCESS in 27m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b1565c9a83e24f88bee9c8e9a93926b8 : SUCCESS in 5m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6cd0a3fc5e6344128b7d0caed2ba5b70 : SUCCESS in 7m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b4e0a6257d742a1ba18283f485395cf : SUCCESS in 10m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3fd58de711084148a728104b83609ca7 : SUCCESS in 21m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aa7d2500371c4a6b97b649d5e43bfeaf : SUCCESS in 22m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d9bfde7ec69443b9875cc3973b2978a : SUCCESS in 49m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/26456a32af6e42d0a480d5de0397c474 : SUCCESS in 1h 20m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b32a9bf1a104f5181a88aaf37170a5d : SUCCESS in 1h 13m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d19086d61e0343dc9e35fc13caf1fa11 : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":1},{"id":"034cac2b2e9f2a8f48be65e805eda2d4e697bb54","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-11 20:49:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2c425a362aa24c94f24983efd21c542503e8e5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 23:28:03.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"dbc3db43130a9ef8d011258ff28df78c63bfdc83","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-12 00:09:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea4c1f185dbb6ca61185c3ca9dd847be0b90b507","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-12 00:10:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"57a46e054a91aa5b40c26d042ca5b13e1fad4e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 01:38:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a544e42dc06c47ed9375fd242f01cf67 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c5a23e8c3244709573e62b0bbebe79 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a11d84d2ff142f2a0c55261a498db3e : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a46969509b9489d99ca1011d9101db5 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89417e93934f4c53b5d1ecdb8638b251 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d592cd1922cb42fd9be7125ca6067866 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/511347970be24075a31fb14cddf1932a : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/497381fe2ef2483996dd4df2727708cb : SUCCESS in 7m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/446ec9f1780a452fb0ae56b3e2d1cdc2 : SUCCESS in 8m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0e23d4fc400846dbad89a7c94f528462 : SUCCESS in 4m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b852fe606fa242d3adb7e6ba6b50e2e7 : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d0a4021b7ed74be89d73a524f1519987 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bdb88f92f8bb4fa5906ffd9e82edf31d : SUCCESS in 12m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/def9f21d0d6a4ed9ad86256daa5fb664 : SUCCESS in 25m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f9781e76fcfa484aa31cc48b7cf4c389 : SUCCESS in 11m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e513fadcb9144ac0a21660049c2280b7 : SUCCESS in 11m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3aa6c68bd1d4c829d4953b05117b10b : SUCCESS in 16m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d8a379ab847b4f08ba25947162ebb5bc : SUCCESS in 4m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a50a7354518d4138b9b277bd425ff4dd : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/568b39562516478aa02025418b5546dd : SUCCESS in 5m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2999492e530248e09538889e589fe123 : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2c61be8edcb8494d96b5051ee95339c2 : SUCCESS in 9m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15b3457fdfba46c4b55126ff5dde93e2 : SUCCESS in 49m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4757a1847d7462cba1c040852cfb185 : SUCCESS in 1h 13m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/040ec70a6a8241c3906e328314b652a4 : SUCCESS in 1h 22m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/02940c0abb70465e93a9b6cc671bea3a : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":2},{"id":"80f90f553df37f4f1ac12dd115c4439bc4263391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 01:38:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"39256be8e49e91ace2d2190eb5d59e0ece56a49c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 03:13:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7bd33e8b1e24c6588dc8c7544267a6a : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d275c9785b4b97830b8a55341f6ce6 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7080cd073d6c4a05ad26b3580b7cbf93 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acd46e41b3204251b1f07f3b3194df5b : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a673f1390bfa4f638bcfb46751e143ac : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed468db1f2034328b245980d0bf1fa2d : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2c559bb630b44164ac72e21bb87bf335 : SUCCESS in 7m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/82f5f91447a749f2862c54335eb3c263 : POST_FAILURE in 8m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2af78e4975be497fae57285a1c0d57c4 : SUCCESS in 4m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ddef0117aba44a5acba719c46f08f6c : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c69c4fcf3f444f39b65c8518922235a7 : SUCCESS in 7m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/22bc4d10dea842ed8416f361fdd8ef03 : SUCCESS in 18m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5c2bec0ec9bd45b7aa49f423631d3700 : SUCCESS in 21m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c249537938f74580bdbd3915fd939ddb : SUCCESS in 11m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/512d3aff63bc4afebe3629ea4bb34fa5 : SUCCESS in 12m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70a36e1e28dd4fa895e030730633a80f : SUCCESS in 14m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8352c360264649e3a1ee123bb4643336 : SUCCESS in 4m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2d79a153c84347849f2bb6c914035fc6 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3bcce25c66674d5abd4ef7e36b178d25 : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8444f1c10e524f649a162581740544a4 : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9eb07b9f67fd46d7a25a44a6c4932356 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c6f03b4860646ceb4aec1146a5c603e : SUCCESS in 48m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e2f12ba771a4a3e98c9bfe207ea790b : SUCCESS in 1h 13m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c248a842fb246c4bb3538b76b916d71 : SUCCESS in 1h 10m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9b575162eeb347f4b352934d484c05aa : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":2},{"id":"11455c979a667511b35d59285fa25d6b0d4a7a97","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-12 03:24:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c120fb64647995c33c87d1bde4e6ffe6fc4499b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 04:51:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b2633dbeb345f6a6e5bbac09261aee : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e64d1cfb5884a3696404c841f85387f : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/269b244bfd784c9f86468162f86a1eae : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bcab5c2a5494242b3a8596da4830ae8 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/296e2e5553a049c79ad5dcc777405cda : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf09fb4bc76d4c23b51ea9c64914dc6a : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f616792f9984b689458be2f95c62b1e : SUCCESS in 5m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eb27833a36ec45b1a04f7624b83936fd : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8bbe16e114f4a568fb46d7c3dda6f5e : SUCCESS in 11m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a8a038eb55a8492db887cea58ecafaad : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/055a2c81ba4142c68b271fbc952c23b6 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/278839e0deb948c2a580ccfbd65e780e : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8356009d39d464183eadd0de0d5d8a1 : SUCCESS in 28m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8bdb7a1e64d642edbcdaf6510e0af13b : SUCCESS in 25m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/acb0d34bad694dfa80c89a88ade58188 : SUCCESS in 14m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a5e96a5a22454c57a7ebe26012f090b8 : SUCCESS in 14m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/719b2872b59747a4bdf9fbc505167120 : SUCCESS in 14m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1f96ed3290c8465cb45ae8fb5a96ff9d : SUCCESS in 4m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/765dcc90c3c542f9b9eee667c231cc2b : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a569a20be6864227b7e7a5f793df2ef6 : SUCCESS in 7m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4047ead440cc4ee794dc5f2d6577730b : SUCCESS in 17m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/db3b2d1d6d24466b9238ba3a22eea930 : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78056f624c544716b28bd1484deac009 : SUCCESS in 52m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e7b9f42921b413598281d97ff277c90 : SUCCESS in 1h 01m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/322437430a1e4ae38f8028b21b19b827 : SUCCESS in 1h 20m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/82d7669bad8d44b682cae05f0887cbda : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8363ea7ca9dff222fe0f0a57c6d43733476e027b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 04:52:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3cc4e63948f5fee361e17d2e0b7ff96762d54ae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 06:14:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd9942571f3b4919854fa5540a2e304d : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8476c99a17b244e585aaf65520bac64e : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f59178271514915b557f275c9008485 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d850ee813c2b413d8f00d3bc05de9411 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbe904bf71274cc783b43ebccc68afa4 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f624b1f75e2e4879ae4ed6a6ac2af1c5 : SUCCESS in 3m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/67fed0251cc9441c9402dca8f6b48ef5 : SUCCESS in 8m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e26d653401eb4399b1d7ac88bf963f74 : SUCCESS in 8m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0f5cffabae6045868f10a09d57249699 : SUCCESS in 4m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ce31694e2d984e7bba455e88c4d37db6 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1a68c67ffb648c9a57efe270f9c4938 : SUCCESS in 6m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f49e191b6524115b19d13a85f429c5e : SUCCESS in 14m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9be5ee823454da98e520bdc29e8b34f : SUCCESS in 23m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5ee353fc2c6b4b0b954442886fc78d80 : SUCCESS in 12m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7eb4e0293790417a86b62c603e1d8469 : SUCCESS in 11m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88a954f9da084b6a8924cae7b51c9adb : SUCCESS in 18m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cc863cd1992243b3bb768a18a80ca343 : SUCCESS in 4m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d7441e6508224523825f0da8d54ee8d7 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55fb4d1cd67c4ba3a965d428e402e6d6 : SUCCESS in 5m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac66e9ab546540fdac08c5c219dcea5d : SUCCESS in 8m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/29f63ac59b044cf1826b84f48c22ed6b : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/020619ca8f37463db2e3c343ad34303d : SUCCESS in 42m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a86053bec8c945aa9f1ee3b8c4ade509 : SUCCESS in 1h 08m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/933a85b1990d4483a074d692c21943f6 : FAILURE in 1h 01m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2e8b4b14cf5e4cffa9c4536495654234 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":2},{"id":"03e6f64bc201f2e32da4a078b46ac7dffa1ac06d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-12 06:40:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ae647e06f19f868806f18ba55dedfbe7fae2b717","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 08:04:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b959b13580b242c7a2845ddbd7ba0992 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a6277a436db4073a830a35e347a8ebe : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97c8b1881e224bc4acdab12f9ad0b7eb : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eed3d42526b04f6c9ca02d8fec84665c : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4922ae39a96483d8b4ecb0efb33a8d6 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a7ee0f02d9f4ddbb7844616d255ab93 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d5727491c2f543e79e27fcdf3f56227a : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e15a7dceebf64eb889b6a51eb1ea1cd5 : SUCCESS in 8m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a71206caae2b45dab0b31a0bf3cf1cb9 : SUCCESS in 11m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/75a2c576629a4c55bfc424a3ba4a180b : SUCCESS in 4m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7478e4beb110428b947acff91f6e3ea9 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab997a10ec394c7588b150995e815f96 : SUCCESS in 6m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb6123d4a7144f2b9ef54d564411a5d8 : SUCCESS in 12m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/826d0e87e1cb47abb0beb73728b03980 : SUCCESS in 27m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f9f2cd02c57248cc85c3398d087be65c : SUCCESS in 14m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ee57331fa72d4e95a65d50aaceb3254b : SUCCESS in 15m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e91612d7122d47dd948f1461f10115b3 : SUCCESS in 16m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b06136b7f0cf4cdbad5c0b379c58132c : SUCCESS in 4m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/09c9f3523078424ea0a7fe8f2ea3b398 : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95ff608a795647b28d5b8dfcc2030115 : SUCCESS in 8m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f5525cd6cd0e435ba6e76f3dfb0b77db : SUCCESS in 10m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ef00897159b424b83905ca46b2a8a83 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a6126e4f961461eb3618436fed5949a : SUCCESS in 58m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e84e5fe05d943c98a38b6814eef3fdc : SUCCESS in 1h 06m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5770110a8a8d4f2bb468f9bccaa83ef4 : SUCCESS in 1h 22m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/33fbc671860f4e5487a2bd897f93ed2c : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":2},{"id":"932f5acfe8852d32de9f45cfa54d364a83ad6134","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 08:04:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d29cd6c4bb6e3eea9093f156adf09fff6dc3f345","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 09:57:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"955b23223b19dc8d050a0a2f7badbccf2684052a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 09:57:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79448005a0494496a481d5ef937d1905 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/398fddeee55744e0932ae63f08f0fba7 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6afdf69e33e04cbc88983062870cda38 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5955edd85a2461ea162b04f4d9e9242 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d92b78488e974c3c971012062c30d7e1 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6e7b5031dd541c6a6433e70848e4eb2 : SUCCESS in 3m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a3c41394700c4241b14f98687fb3262d : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d05abd4abd341aab74246217b2cf9c4 : SUCCESS in 7m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/87410df55673433fa0d600450bf1035f : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ca25ced251042b99ff59ccb50c1c183 : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca87f7244c5e47c38f0c322c25cfd810 : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5ccbe104420b460bb0b603d980da94a8 : SUCCESS in 14m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/92f855da937d486fb70854fdf59a7080 : SUCCESS in 28m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/25a9a8f6f848428e8d7fb4981e5f49c7 : SUCCESS in 13m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c751baf384aa4714bdf67572f1783536 : SUCCESS in 13m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5956f9045a4e4560b6cd532f823f98c5 : SUCCESS in 16m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b005f436d5b24a25a358fd7fb15cd26d : SUCCESS in 4m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f1bc8bbb72924db3b35f566eeb36c268 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dfb899bdeb944562ac6166d3bef7c9a3 : SUCCESS in 6m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/effef78521b84e8790fd478a45d23ddb : SUCCESS in 10m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba56bb18c2ab4dd799f1d48f694d62c7 : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46859bdbb69e44f2a31c47c7994319f8 : SUCCESS in 45m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/65e4ccbf5a5b4137b706416fcc18d403 : SUCCESS in 1h 19m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05008eb7abb14ac89620f9dc9942dc07 : SUCCESS in 1h 31m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4de5938eeaed4485b1e4255990f65728 : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":2},{"id":"616a575af562b93c95632cedafeaba3959ea2505","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 09:58:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2af500e84a84a7a8b41386f812819ef : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"917541afa0cd9f2fff6b681e416569a2ddaaa178","revisions":{"a372f503ca0b9146a98c39edbe2625e8a4e8e7a5":{"kind":"REWORK","_number":1,"created":"2020-04-11 19:09:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/719306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/719306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/719306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/719306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/719306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/719306/1"}}},"commit":{"parents":[{"commit":"8a193ffdc85b6fbb189b469cdbbad351d9ef8939","subject":"Merge \"update constraint for python-heatclient to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a193ffdc85b6fbb189b469cdbbad351d9ef8939"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-11 19:09:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-11 19:09:25.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 4.0.1","message":"update constraint for python-zunclient to new release 4.0.1\n\nChange-Id: I5ef713d22bbe570ae9c78ac00dba5fe0b370c17e\nmeta:version: 4.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Commit: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Change-Id: I7adda48a2fbd2fea9bf80d9a4264cf79180c2734\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a372f503ca0b9146a98c39edbe2625e8a4e8e7a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a372f503ca0b9146a98c39edbe2625e8a4e8e7a5"}]},"branch":"refs/heads/master"},"917541afa0cd9f2fff6b681e416569a2ddaaa178":{"kind":"REWORK","_number":2,"created":"2020-04-12 00:09:58.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/06/719306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/719306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/719306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/719306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/719306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/719306/2"}}},"commit":{"parents":[{"commit":"1815927b33d521da5369d446972fb05aa65df50c","subject":"Merge \"update constraint for python-novaclient to new release 17.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1815927b33d521da5369d446972fb05aa65df50c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-11 19:09:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-04-12 00:09:45.000000000","tz":-300},"subject":"update constraint for python-zunclient to new release 4.0.1","message":"update constraint for python-zunclient to new release 4.0.1\n\nChange-Id: I5ef713d22bbe570ae9c78ac00dba5fe0b370c17e\nmeta:version: 4.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Commit: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta:release:Change-Id: I7adda48a2fbd2fea9bf80d9a4264cf79180c2734\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/917541afa0cd9f2fff6b681e416569a2ddaaa178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/917541afa0cd9f2fff6b681e416569a2ddaaa178"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
