)]}'
{"id":"openstack%2Frequirements~719345","triplet_id":"openstack%2Frequirements~master~Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-04-12 16:22:24.000000000","updated":"2020-05-03 04:27:36.000000000","submitted":"2020-05-03 04:26:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719345-1588479985680-3c423189","meta_rev_id":"0d0718faa289cbd3aad68e7bc2610bea3e705fd0","_number":719345,"virtual_id_number":719345,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-03 04:26:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-02 23:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-03 03:08:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 06:19:00.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-03 03:08:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-03 04:26:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afc3969b1129b5fb0a60712b65a564f995a4044c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-12 16:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae2e068029cff79472a6da3371a2f212dc7b5a79","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-12 16:42:00.000000000","message":"Patch Set 1: Workflow-1\n\nCHILL","accounts_in_message":[],"_revision_number":1},{"id":"084fa3fc1f53c8ccfe6668bb5d7210f64d9803d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 17:48:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cad2bc3a224ff9bf5dae555a7e409f : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43603ffd4d04cbaa7d795e84f171847 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2290121442774a829b39370cf9d83081 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b28f0837ca8a4c0bba025d46203c8b9f : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd583d18e95448eab5d8ccdf931e084e : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b046cde4afa44d1f8897d7bd095387c9 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b564ab2d684459bb734a4c27de959c5 : SUCCESS in 2m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd6a0d3eac274d33a250db51b3f97d5b : FAILURE in 6m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/91f1f7f3dc7e4dc6bfd0da1160e13f81 : SUCCESS in 8m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed4457b42dcb44c788c3791a4b711ed1 : SUCCESS in 4m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0da75fe7a86f480db9eaa539b03484f8 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6e9625b8f7f4ba38d9d90f54b730db2 : SUCCESS in 6m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b6bbfc4090174749851514f4f767c96f : SUCCESS in 11m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f05d8a2f11244f15b556b7dc9824d8b9 : FAILURE in 25m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c20fabcf5b5b4b65b1c70b2072d50c75 : FAILURE in 11m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6f37e1d4b24f4aa7a327ccb7ce316c82 : FAILURE in 14m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3820f45b1f04281b34ad7d7e83829d0 : SUCCESS in 15m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6273a4c476be4bd18ef194ab4c9e298e : SUCCESS in 4m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/62d7917a698242c9800f84aa2a89ac64 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b0895ba105f47e2b21ce7b6a323ff87 : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/26fb848ce70e4615aace1dbc41c9527f : SUCCESS in 9m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f18c21d513c741fb8dc6dcbab31f5858 : SUCCESS in 9m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb0cabd42a844d6f86fd40f2323e74b2 : SUCCESS in 50m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ba8e25368704724aca19032796b1821 : FAILURE in 1h 15m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/010174e96dcd41bca5d9539480b403e4 : SUCCESS in 1h 14m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5e2cdc72ec4f41699131d8ad3247d0a7 : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"cc9198dc5e106d3c9f4f14023064e4e75633c8f6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-13 06:13:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"494769099486cc9e04f72ec775120b810ee91e64","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-13 06:34:44.000000000","message":"Patch Set 2: Code-Review-2 Workflow-1\n\nFREEZE","accounts_in_message":[],"_revision_number":2},{"id":"1d44b25d937ff98e281081b79b70ee9f4530e7d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 08:09:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120319f1bb8d484d93710d0a80222bf7 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f166c69a96cf47c69a7b221b482846be : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ebbe7b2500e40f3981c5ad1026b3508 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57fced875d464dff8af328a262f30812 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5514d9cc3eaf4bf1a4b778b9149eeec4 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d0a6d3ce36a48db9c5b16b3b04e087a : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83a9cb1ff52e4b048dbfd8a9485b3595 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2f9253c507f4d5fba9cf8b1e38ef9a6 : FAILURE in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a15eeac3a6a44f41af65bf315efe3625 : SUCCESS in 8m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ebd2d10e7c3a440f856f8d1b8e992216 : SUCCESS in 5m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f9469d7d86634c299a7cb3387fbe778e : SUCCESS in 8m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c3b22fbfa470440ebda3efcf9845ac4d : SUCCESS in 8m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7b336d9e57bd4434b1c15aa05b1fa370 : SUCCESS in 23m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2b59024a852a4171b51be6da7385cab1 : FAILURE in 25m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6f9647664f964f06a99d93fe27c69b9d : FAILURE in 11m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cfb4873f0cce4e7c8f8f4a2b78f453af : FAILURE in 12m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f70d76e10bb472ca7f80eebca099a03 : SUCCESS in 17m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d240ec62f34e410e9712eb4ba93a7550 : SUCCESS in 8m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c4c965fb684b43068d370b1eae1a771a : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d13a2fafca1147cc8d14200d958c5fb1 : SUCCESS in 6m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ae4bbf2b8ea14ba19cc7613124007154 : SUCCESS in 9m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6df6fa6fb3534f70b043c14dbd0f9e01 : SUCCESS in 15m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e8db57cce0b40299860dbbd67a2a752 : SUCCESS in 52m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/66db93fc88554a8b87a35f43445bcdb8 : FAILURE in 1h 15m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d67868c17be54dba9583fc8d8a3ec828 : SUCCESS in 1h 20m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c4a25b743c8a44a895ca7afaab0660a3 : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":2},{"id":"c4468dd41d96b2635bf5a1ca816baa3bf1f095ae","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-14 06:16:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"faafc84f57548f4dda41021a2d82bec1e91a14de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 08:10:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf11d5ecbccd4c8b817e20a4e05e8949 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51d11c0176d4590a4fe389d4f901a60 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffc1dae83c1d4ec8ae2b409c83aa4b3e : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/335312a8978b46038869149ff3391254 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc6e73a74ba346ecb55df23587d0109b : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7dd24139c85a4e12af022d621e546a18 : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aeaafb53b19343a8b394760ae6d75224 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3c7fc308cea6444aad99adda1a19aa7e : FAILURE in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7413073e4307449bbf0ba4ce047fe369 : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2943d1117b2c47e691f81394586b4c90 : SUCCESS in 8m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fc9c3323d8d946d4b2321ad7dc1f2473 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e169ad2bc7f4cc5b31c03e40d6a640a : SUCCESS in 8m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0d204f8b31374634a0f3cbf23cd39678 : SUCCESS in 30m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6b66777a6f0b4e1fb73a9326d180cbaa : FAILURE in 33m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/07df20f7c94e4bb7a05968d54919615f : FAILURE in 14m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/641ad8edf79f401196d30af23b8bffeb : FAILURE in 16m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fc90b4f16bb48b2a82f9663b294536a : SUCCESS in 16m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/edad17696afa40b88c2ebf3f9bab5c94 : SUCCESS in 4m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cc637819d8dc438e9b0266020d4250e5 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e812fc926024e87af324d0d5b6a2c8f : SUCCESS in 7m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/216d28641e48442fbca3562911693761 : SUCCESS in 10m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fba67db90d73425ea43506ff8d3bc742 : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca808e761ec145e48db7eefa9aab2e04 : SUCCESS in 52m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36f8dbea2bcc47e29d40e3f66923f0e9 : FAILURE in 1h 09m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba3c8054a36b49328a9a6e689208c036 : SUCCESS in 1h 17m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/45db927a474441db9326ef904bd11b75 : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":3},{"id":"f5e7fd4eec6a476d7c63c24688b644a04b81447e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-15 06:20:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6540c7e60868bebfa70175e016a7d2cbaba06efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 08:24:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c7ff7880efe47cf9b6655a324846e6d : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf029014dd484855835ec38a8975fa74 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd9d0c094864775b36964eded69ea6e : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25694e612be8407680eaebe7ba31ffa1 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4460da5cc2c7423bafc053743bb56f19 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3be8d919f2b34c42ab0e3340efb15b13 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4069a57bb15f4d5781a9cc64ad109019 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/298d8ffceccf430185f2dafce9d8c384 : FAILURE in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5566dc61836a4fedbe0b0a8aaccfcb96 : SUCCESS in 11m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5f8eac6afddd4b03b54ff5740662bf66 : SUCCESS in 6m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bf4b6cd7433340da820d15c6841f558c : SUCCESS in 13m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/697757d57e2d4baa806bf60e43df8962 : SUCCESS in 9m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8249c4b9cd54e26a29c9c5c6512b3ca : SUCCESS in 15m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cad3e4ce77f14b4ba1abbc69a6e74dca : FAILURE in 27m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/08ffcc80e67b43438d012002fd96a2fc : FAILURE in 14m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/036d4f84f9f24349815ef6cc367bf62c : FAILURE in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a221410c4564481adbffa2217b6872f : SUCCESS in 20m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3a7dfef3cc2c4edfb4ecd7aa5d9dd911 : SUCCESS in 7m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/930bb54036cb4660b830c1cf019b79d5 : SUCCESS in 7m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03385ce64e934ef4bc310c67bdbb156c : SUCCESS in 9m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/85502a11b58a4e86a73a4cc186373ddd : SUCCESS in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7d6c7fcab9a9430f9190b870ddd39492 : SUCCESS in 11m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80f243b4f0a94867a228cff5bec79f6b : SUCCESS in 59m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/539d58bf070948b798dd5679b20eecad : FAILURE in 1h 14m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fa7a71ffeac450bacc4c0c486b2c047 : SUCCESS in 1h 19m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b9f89ecae3964db6b0fda6b235b4f6f4 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":4},{"id":"afcf31611399ff6f9147315834eede9ed5cd1efc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-17 06:16:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c0163b3f39c9c2c227d78c9e23634d25b0d6566e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 08:02:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0a6ee41a994c63a146cc7b47e65e3f : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8631796e2c2e4cf88bb67eb2aeb6c6b3 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22994d22b5b14e35a545ffab7eecc1f0 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a231d77c15b498ab03e883f2bfedee5 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59bd9e9a7d954d41a53f5ba72b8494df : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0506d6653a54b43b3f6b68b3614119b : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1893bd009e3946afa572cfa338630ffc : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c24210f9bc914b5596e9c9f66e8b1bb9 : FAILURE in 11m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/15ff85d993f7493caf0e80c46a6f8540 : SUCCESS in 9m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b9276fbc02bf4791a2896895b262564b : SUCCESS in 13m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48757a53adad47e8a48de50a2ec82f46 : SUCCESS in 12m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8d25b4619424eccb5b2b290767c7bdc : SUCCESS in 10m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/696d8fcdac8a49f384f9e286dc74c02f : SUCCESS in 24m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3508383280f34675b8206e1b182f6f8a : FAILURE in 32m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e6728bef904b458ead75500a3577dc7a : FAILURE in 16m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6508ba346a01478babef02497582bc4f : FAILURE in 14m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52cb261ab02c403f937d77d342c570e8 : SUCCESS in 20m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0b5c5a5b887145dba54aecd6c7be6e66 : SUCCESS in 7m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/62404dc984544e7f956ee395265afbfb : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0432ab5752374ae8ade4b3eab6da19ec : SUCCESS in 8m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c034ad7ce5c41aba6eaead55595f420 : SUCCESS in 11m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b8fc6f2a609f485e810c0864c51b350b : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07ee3a23e0fd4d96b5bb9107a9a5d3b5 : SUCCESS in 53m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/473b5a34df114e2596c5bf79ac0ffe77 : FAILURE in 1h 22m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5bf80f3170f4c8b8570f97f04171829 : SUCCESS in 1h 25m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/826e48fd185d4f5bb0cee715d1830f87 : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":5},{"id":"f8601147e839ba84545cca775647966dc09c3501","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-18 06:21:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"68d6238849527e7d392f3cdfcc537bd0aeab8110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 07:52:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56410f04a234273b8df3769d7ccf519 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b8bded0ca9f4206a5686a7851b0b8e5 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0038d0847959479abb209e591d1b29a4 : SUCCESS in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c792e72cc7f4965a4e88381362a7cbf : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a1ed2c259d34510af1881c1e2305def : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/756479720a9e4e35854b3e377af9829d : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08cd3167d20c449294610a61aeb29993 : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8aa637c839b84c26bccef6f4a768141e : FAILURE in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/20f6362802a744dbb656ffbf4e66a36c : SUCCESS in 12m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eb5b668941604937801c6325869ddf40 : SUCCESS in 6m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e88c9b47791c4eafb29fc64a4ebc12c3 : SUCCESS in 12m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d9cc6a49932452798b22b1a4ad8c00f : SUCCESS in 10m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/08ca4d3ac7da4f5b86613e35696f73ab : SUCCESS in 14m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/df614a1f8ff14a6385c7a80b36ab2fde : FAILURE in 28m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f052a4a405c34c12a731c7f7b69dae0a : FAILURE in 15m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ff9a37fedccb45c28a369cc75d4ec95a : FAILURE in 15m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/270b6d88f78a42a08e0c979d829cd313 : SUCCESS in 20m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ce8ddb8715e94a60a5e939eeea4f1ac7 : SUCCESS in 7m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4cacf22aec1e4c199ba20df4d7cd18ae : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77c0800dd28d4d76b0595e15aed6b72d : SUCCESS in 8m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ccf8ab5e9e26439ea6d9eb80d49a02b8 : SUCCESS in 13m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/825722686e3d4f7096f90bf650d93ee7 : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e99583f5ad794aa2aabdf8a91ae9aa6d : SUCCESS in 52m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b38e24c9a4644843ae4daaaa86723514 : FAILURE in 1h 22m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c1bf7fa5c8c4c53b3d64308b6cfd18f : SUCCESS in 1h 20m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/792910aaa19042ad8a055b99f4d74420 : SUCCESS in 1h 12m 53s","accounts_in_message":[],"_revision_number":6},{"id":"6139fa314037ab0c2dbafc9387a0c6c654c14d67","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-19 06:22:33.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c4759c181deeee130ca5ebc8485011ebee1a9040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 08:00:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662b6ed830b442408741ed824967acd6 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f68ffebed0463194185c938d4fc0fe : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecbcd33863aa4757bbfd6fe67b71c1cc : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/335f6024fc044135a23fe48c3bfcbf52 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39d56eb4fbd945f790b745ffae20c461 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c10fa4d3e7074d6aabbd3ee50657707a : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3aebab0df1564aa98b9c047fc47401ae : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/729c834ff91b4de7936c36071ebaf647 : FAILURE in 15m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/57461e1136fe461aaa6dea0b20307f11 : SUCCESS in 12m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3cfb3d8106d4a369c7d1f8c6b4d9bb6 : SUCCESS in 7m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/24437545a63a4631b7434d648bb064be : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d97202c887e14a64b647465943561191 : SUCCESS in 7m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fcc69e4c25bf40a9a765841555ca3290 : SUCCESS in 13m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee28e8525c0248f6b09e67132b247642 : FAILURE in 25m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5aff50eef5214f4dac05be0c23f4edc3 : FAILURE in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1b2b82cbf2e34d598e9d0288a744f46b : FAILURE in 17m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d4fe0514a7a4870be983fd4d1f340ed : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7e1c78025d074ceeaf639713539abce0 : SUCCESS in 6m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d45f589d2fb84c729958daeee404dcda : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9da91104f7704243adc5f18a484e4ad4 : SUCCESS in 7m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/675a4f3a25de443ebcc6d613fcfaf1f8 : SUCCESS in 20m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a9041355a8184c8c8430273943dbf72e : SUCCESS in 9m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a3079f7af174be1bfbc494c6abc3dfd : SUCCESS in 51m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78759e0012a9415a92a9e5648085f9c0 : FAILURE in 1h 20m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c47f21652f214581a22c32601d16d1d9 : SUCCESS in 1h 22m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/526eff483de645e5a7b0b312bc8d868b : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":7},{"id":"7e566705fa1660191a70e37590b477c1fe9c8b06","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-20 06:19:17.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b30a26959f8cfa811bf707c9dadd3fabc0a2173e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 08:09:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc157ccb21fe45c7927d3843512c3ef5 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a065e6b596494092c1bcd9afc75f36 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaa5ae99c15e45429d6d28e779836305 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7ae9dfcf77e45f1a367243392c56580 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66e0029d8e4c4c4d850ece411400cdc3 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/049608a78aca408cb93250c2ee11528f : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93c60023adc34c2a8bade3ef63706d98 : SUCCESS in 6m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f584fce8c1e045a49c5835b0ab4b0552 : FAILURE in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9c4a332442974c9a96ccd006765f6df6 : SUCCESS in 9m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f339b86f19c44ea0b374559c6981c917 : SUCCESS in 7m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ce6bb7acc5746df97a8290d11064c12 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9ec1ad84c5c4811acd556be679ff56d : SUCCESS in 8m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3d524cc86cb043bb80435b2324083142 : SUCCESS in 15m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/616f36cd11d24f2688625f55cfd52bec : FAILURE in 29m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1eb2a526b31944d6a20945ebfa0e9dfd : FAILURE in 14m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e0573eecdb2447a4bc924031ad5fb23d : FAILURE in 15m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/191f9d7e2c66468d983606b587d34d8f : SUCCESS in 17m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8c9334720c74e2f81fb3f8ce2ff1c8b : SUCCESS in 6m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b461b48783844054a15bc63f94f32817 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/acf2a18780e24c4c90aa463d8297e3c4 : SUCCESS in 9m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c8f75540a2674b0eb6ac4954fc99ba7f : SUCCESS in 19m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a6c3ab233814023a42198d0e8e5f10d : SUCCESS in 18m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa403eb110fa45289b183a1e6067bee6 : SUCCESS in 56m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0fe5e493c624296a2eca736a08975df : FAILURE in 1h 07m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab3b6ae9c85c4e66ba9cceb0fec39aba : SUCCESS in 1h 22m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/331705c91d4242c2ba2d7047493489eb : SUCCESS in 1h 14m 54s","accounts_in_message":[],"_revision_number":8},{"id":"c8d29eb3739d5752220409c8ab8da299fecdebc1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-21 06:15:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1c2afc63dc7d1175222b0a7890392b2d11c617ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:26:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0804fdc3e5497eaf3d19c205808d46 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efedf7b7c3ba46a79840decc05b21c9b : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/903b067961e04e9288d66249a6809316 : SUCCESS in 8m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4307ae99fb8f431bb1b45f21313f3aad : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eea27151fc544cefb51bd13c9b537b26 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a2acd124f764ccc9b324d0032ceb37f : SUCCESS in 6m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bf646dacdd24dfdb3d51d88d7beffda : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8789c12dacb04ab8bff0b5768213bbf5 : FAILURE in 11m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e88f3fff31834a35b53123600332fd3f : FAILURE in 10m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3af4d1afbc244a06ae5e70884e4fac8d : SUCCESS in 7m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/83b7d8a36a614d6792307046e7148498 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0375e7255e694afb859c80b09b39ffbf : SUCCESS in 10m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ae358a23568c4ea9a0b4c88c3da6965e : SUCCESS in 23m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3df9efb09fe54627b6af7fe09261217e : FAILURE in 27m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8372f8f24f4e4ee88dfc42476ac4858c : FAILURE in 17m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5b7f6fd6fdea49298ec8a62486a68fcf : FAILURE in 15m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4871b0eff524ef29406737311b1de0d : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/36c5685800e44e309d480fdb8a878143 : SUCCESS in 8m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/961c6edb7f0f44eeb1099be9bac17109 : SUCCESS in 7m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d53856efed234c0fbcd83924e6aa4e42 : SUCCESS in 10m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2f521c177d1340d49001f38c99bed58b : SUCCESS in 15m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a79f3aca76f64247ab233db5b3522a1f : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1e744403e1c4b1fb522f9523bd46c4b : SUCCESS in 51m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1a756773f7f42dba52eb795c4534dbb : FAILURE in 1h 23m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce341954bc2b4608ae42ee6c417c02e5 : SUCCESS in 1h 27m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2c707a5ba56f4423b260699713421157 : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":9},{"id":"a9b791efe0fd1a6e1dcc8685139af7eba7b0cd31","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-22 06:13:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2c1d7603aacdc7a12969769b2932649d530c2487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 08:44:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90decfd12ce64035a003a1147bc77ee3 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c8c410448440469c45884aa8400fe0 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51b5d49e96b94a55a97ab4febddaf917 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f5bc0e0d39d4efd83a204692179a5af : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59175f00dc1c4f04846c5edbf072eda3 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08b731b164c9415b894f660a5b292803 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96f8cd8e6f654e0a886fc1a6a4bac94e : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1c906f2c120342399b4c0a6d1b4cdde0 : FAILURE in 7m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/454c2faddb98409181b55803bd962bc6 : FAILURE in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d9ef9f8260e54f159d73539c29f06940 : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/effad232202f49138b795bc174dcdbb9 : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3a88eaab43542e8855c59a986d83487 : SUCCESS in 7m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6052f0aa29f143188668f1374752ded3 : SUCCESS in 25m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2b86ed90b535423e9c92a5d18944136a : FAILURE in 29m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ee5e2ed53ba644b6b988b15c4df2a837 : FAILURE in 13m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8e1b1ae2dc534f2a9254bf5997f0309f : FAILURE in 13m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12c6e89c46914ce3b69ac48967393f73 : SUCCESS in 17m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/920d6e5445dd468fadbdca3323f58d68 : SUCCESS in 4m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/14a3e4393cb44e3ba79384374f6a925f : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f8b5d64f4e0463683b5b0b5537037bc : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cb59198aec5f44d4ac4b8ffd2773e01a : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/283f6f15ab7b4d0da55f849c485c0504 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5220959d8b8743adb8fe8cd666469f72 : SUCCESS in 58m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f50a3d17ce9744698537342bed402c1f : FAILURE in 1h 15m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e02d6b1e12024133a05e19a7d2769383 : SUCCESS in 1h 27m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6702febae686483d945937a5ef7124b0 : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":10},{"id":"d70f0299ab91f7fbe551528ff3e602b9a334424e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-23 06:14:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"690b25cf8c11af2bbc4c03962023e47f59eae9f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 08:50:45.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca699f33f934ee69af4c76209f5abb8 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9971909fcb483991db63262e1a66f5 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee68640a7a1e4c98b5f8929a20cac42e : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f5ace67f78e4c5f96fdf4b5185919e4 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5569f2cf45f4e5a8aebc5d38d401a96 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e791df800f4437589465353a6ec742b : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5afc2040bfbf4e10aa587408e070cf2a : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/30003b02312a43d7a29ac15cfc43fd12 : FAILURE in 11m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b4aa829f05ab4e0c94461fdacde132b0 : FAILURE in 12m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/404bef2f071943f0b71dad099a8a12af : SUCCESS in 8m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6e5fa856ff294ad589f890146bb6d605 : SUCCESS in 13m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80738ae21a79406baa81d4531646f1aa : SUCCESS in 10m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/63a86f9590a2473a8356e1d3f8abc90c : SUCCESS in 25m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5b034d902c82472a92369b7b629b2228 : FAILURE in 29m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d76c5bdbcd5b4ff099a16b0573972d16 : FAILURE in 18m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/df97fff0369c4e18b89242118e61d423 : FAILURE in 17m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/663d1a1ad7ae4ee2877334756b4f6028 : SUCCESS in 20m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/83ff37c829dd479f84e06041f52aa61f : SUCCESS in 5m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/541d9de9aff64916b0f6c3348dfbf9c0 : SUCCESS in 7m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f7e0248c5f347afa48c83f82078bfac : SUCCESS in 8m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c4224cfd0270465e8b276ce92309c099 : SUCCESS in 13m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/39437ecf150d41c79ded26a41cd2da98 : SUCCESS in 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfa4cac01e5943698047eedb9e1a41da : SUCCESS in 53m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcaf2267be0341a3a896db17ae3d68ef : FAILURE in 1h 30m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c7d336bdf0d4947b13b030a73013f85 : SUCCESS in 1h 15m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/443d96d1755642f1ae27cc6c62673a3d : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":11},{"id":"f5e6973f97539c19ba449f29404949488a4680f4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-24 06:13:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2665b4aff31b8775e1378012d8ba38bdc33b26fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 08:12:06.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ba4404346b4c2cb8f6033fdc7e1277 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7dadc0e008348d8a66e94b5096486ae : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db966599daa8406ba451ef35864b2830 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38648f482b4c49d1b121b1ec3aeea767 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba061d69e4a64818a92838445bd09f42 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4cac4cc428e34b18a96c54587ed21472 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d278ff293924269873d4d4ecb741f91 : SUCCESS in 5m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8fe3f0ac3f2041f68e113eb19ee0b442 : FAILURE in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/689018419a99420db61d73e67295b166 : FAILURE in 11m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e5b048d5b6c842ceb07a4c9455d0675f : SUCCESS in 7m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ebd8172d7b57452c8d8ff7e6e024aa48 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73d1586895234a7fb4f6842c8f42bfbe : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7aa03bd08cfc40989077fa61f1714976 : SUCCESS in 16m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/386c2dc55e504b82b5cb588bea43159c : FAILURE in 31m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0769bc826e9b47808ca198262dec6b43 : FAILURE in 16m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ed45cc0a2554460fb787edcc12d14648 : FAILURE in 15m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1721cbd01a2473592e26157acc65905 : SUCCESS in 19m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/09da38eb5e0348c7be7d3b02f313ebab : SUCCESS in 6m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/90f2203fe24541bc9c5e6a9c013ec833 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7d461abbe5e461b949194462377fbfc : SUCCESS in 7m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6aceadde5f2e4aec978824112650927a : SUCCESS in 11m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/96f761eaf8a74b159e048fd116994c22 : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/857c457bdf72449b9990924764770445 : SUCCESS in 56m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cc5347d2060470c931fed1a8d20df53 : FAILURE in 1h 21m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ca73f76a24c4305bc81feaec10cd2e8 : SUCCESS in 1h 23m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/52bfeaab4b134133b934612d4bc9e898 : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":12},{"id":"c24eb92b1cd0022f74ceaddb6d32b0bbe94c6b6c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-25 06:45:46.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5922135b0f2b47d8a27ec9a5296693e2e2c6915a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 08:50:33.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876634f14d4e4f86be761ff4d7bbb54d : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98911c6e00d40b6a637b49548def469 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57634c5011244604b9e2cceec0dd754b : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa0f3da587ea4db2b514216b9c5c4e48 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7588ca9e473482cb03f448621bfa2d6 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9ce3f989893437aa39a32457817c117 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/656bb2304a734e12b486e2f71b27fb5c : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/403d58a7e45b493a9a30f6502ec98edc : FAILURE in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/41067eb8bd8548faa6bcbb3c603176f1 : FAILURE in 12m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8dce69a4e15948a3bc020f470f8f0af7 : SUCCESS in 7m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/56bf38483ef24e94813871aa8df15f60 : SUCCESS in 13m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f41bd55a0d3f4eea98ea0eea6be01cd5 : SUCCESS in 10m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/18836d2f10e54834bc869ebfd54e5ee3 : SUCCESS in 20m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf39b145bf4c408183558153a6b42e00 : FAILURE in 23m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b208e413beae45c790c7544c3d432e26 : FAILURE in 15m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/67de4ebe8026444e8374c0fb5fe94688 : FAILURE in 17m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54a8124c5c494130ab262f636f31e5d1 : SUCCESS in 20m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0ac69e7d5e574f0495021345da6fd604 : SUCCESS in 7m 35s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e2ff4f2524ce4f229d5e711db0105973 : SUCCESS in 8m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c6948873eb94ebc94361e749fe4738f : SUCCESS in 9m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/62625c0469204ed397ac2bd692a5f01f : SUCCESS in 13m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/054009d613744d19b62f6ca538ea3c6a : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/124de79e39b84e91bb060ab9662df21f : SUCCESS in 56m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd251578550e4bf79b74993c2a97d807 : FAILURE in 1h 10m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc627b71c62c4704ab5270cc9ce967ce : SUCCESS in 1h 35m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6acce7d752834b19a0dcd1899d4e06d4 : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":13},{"id":"8bc80602e734d4a1e476bef027e1aa34c8f0c7e4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-26 06:13:23.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5da3b3b8a4c7ac4bc68da9a1704281a2d8f3cfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 07:54:38.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c170a873a88f4a05a7ddce08ba30e433 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb849e4863a4aaa9b62cb4c6d72de12 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f6541564d5d49b180a31783adeb3ef6 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c042e54b28145a5aec37ecfe38bd19b : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ef8a1df9d464630b2c96e39df5f8aab : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6ffa5b7e45a43a0aac918fdde6cd190 : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2500e341eacb4d259777ae5e5e4e0938 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a20a9d3d0f34574af2d542721d1eb01 : FAILURE in 11m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0573f02661d8459582a86ea00fbeeef8 : FAILURE in 14m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/797f59a27cc04274a17625f359b95f30 : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5553f75336214bc9a9d2229d87c0e777 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ca439cdbfd84bc3af3e14eadfc7a363 : SUCCESS in 9m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5aeacb9a49fe48d6a284c1dbbd4e39cd : SUCCESS in 31m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bec1190c125f449f92e069e21515ceab : FAILURE in 34m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46028ad24f5c478894f8c29883216e45 : FAILURE in 17m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d8efb7a5f0d548a2a61164a498475b38 : FAILURE in 15m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b70f96f158a4e1ba92cd1359c6466dd : SUCCESS in 19m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a405651a98414ff49848940f8a713a8f : SUCCESS in 8m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ea7f189bc27f4142a3efa8c597adb1c3 : SUCCESS in 9m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/855e83d3df534440bf329e2def566895 : SUCCESS in 8m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/99c94d0de9f94fd4800caf8da253b477 : SUCCESS in 14m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bef7e6302fd444b0bbdb1b8f8a724e3b : SUCCESS in 14m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9374c1e3d3d242b7b6de6f4fb13a449f : SUCCESS in 57m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a366ce2512442409840e9d600dcae1b : FAILURE in 1h 27m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f176cdb2e54489ea091b4f2ee31323a : SUCCESS in 1h 20m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/15349643ffd546dc877db69577639424 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":14},{"id":"a858de27a373c9c08d5199cc862fb1cbddf2050f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-27 06:59:58.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"9c77ff0b812fcdf7d45ce1dc7bbc7c5de1ced6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:32:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f26c14db6e5d4e179d85f2ec374fb7f3 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57825de49b6a4cb29740dd90af2511a5 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/754fbd7fd3fa486b9571700f55b935e9 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/693bcf31f3b1467d99a93d3bb61defd1 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5a49cd725324fa0b2b2ac68a28ba3e4 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8a25595b91d40059e3cc9b3a97eae91 : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09731bb6e0d5421296888641ed305cc8 : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/68fe038e1d46426fa3e95352d3c6f99a : FAILURE in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6bd0fca4b3af4a75a5ebb707b8e2972b : FAILURE in 9m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7ec6d5d7ede640e6bc0c8725a5d403c5 : SUCCESS in 6m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/96d396a538834308bbf2a821418f3d1f : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f329dea3c9d4188932916ae69d14faf : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c5baa15a75164970babbea96f8960025 : SUCCESS in 14m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/051568f92dd743a8ad4303b90a7717ea : FAILURE in 28m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c160034384cf40b7aedc8f87d91cf5bd : FAILURE in 13m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ddc9cab91f59455ea0ea85fdf0550b9b : FAILURE in 13m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec5f065b2f984a918413976b325b14bc : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cdd000027cf0479084fff1542e985610 : SUCCESS in 7m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/633173f76dbe457b89528848a8c1196e : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d28921df532e43689c142bebb5c3a060 : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/78a48a0dfb9e4ca081724907f4f2f5ca : SUCCESS in 11m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/81dbaadd0dd44094b5bf6091dbd19bb8 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c822ef0f5e244c8bbaf81f7de186120 : SUCCESS in 54m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab217b696ef248babaad9d4661b81407 : FAILURE in 1h 22m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02b353d8183c40fd9224c6ad32d10906 : SUCCESS in 1h 19m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/342453e20e744611a31c4b44e5428844 : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":15},{"id":"87c5a8c5ad221f597285c3942aa53bbb943efbd3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-28 06:15:14.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"70ef7f405424d05a691318fccbac9cffc701919b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:35:57.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936713d84a354111bfdb62a580b8caac : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24fb835be5746509d10f3d9423cf240 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1222ff6244b48bea6ea58a615f1de3a : SUCCESS in 9m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/541bc4e5b9c74602bc0f8cbec1d25bbc : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8668923d6a34b26bb87825db7afb5d3 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ee24c2ea3ec4895bc5a8c4ba3b94224 : SUCCESS in 9m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a9d86543c8c4c80b7104d1cff3965ef : SUCCESS in 8m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ee74e06ede68404091ea8e7bd8a2666b : FAILURE in 14m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4166081367af441d8079924dcad9bb6b : FAILURE in 17m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1215acf3c8b947cfb5af57abf59dacca : SUCCESS in 8m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/04089b7edc3f44cfbf048a07b1350021 : SUCCESS in 13m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d07ffd88e9ec4c598f02ed0876c1c686 : SUCCESS in 11m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/79d983b254e242ed961abe59cdcf49f1 : SUCCESS in 13m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fac3fde2c94a475cb09197efe6f510f1 : FAILURE in 24m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aeb07498ae954cb781d03638f6b8b7a4 : FAILURE in 15m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5f0e73cf699c412ebf41e46326d64b3d : FAILURE in 16m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/367352f3a74b4e98afc197372db24a75 : SUCCESS in 17m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2ffe0af150ac454c861aa1d5f3b88788 : SUCCESS in 7m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ec85f8972cfb47e6847b645a800f1792 : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e8e316450dc4eba8fb67e7facf54edd : SUCCESS in 8m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7ed7133b902a42c4b10f179f599147d0 : SUCCESS in 12m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c2a870182e90409ea8a7c967ef52d2cd : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16dd7bfe4086446688f5aa7b8a56ed9b : SUCCESS in 55m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef4c41040ab04aa6b5c67c5815561cdd : FAILURE in 1h 26m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86a4a90dd2b9452b9c555cc3def93eba : SUCCESS in 1h 22m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/458755dfd5214cc08c5aecac5a70990b : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":16},{"id":"1a49aa19585ad5b075932805b085d5ecbb68ea50","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-29 06:22:08.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"cf95f58acc043cd768761e7a87c97029ec1af3c0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-29 06:22:33.000000000","message":"Patch Set 17:\n\nUnable to freeze job graph: The nodeset \"ubuntu-bionic\" was not found.","accounts_in_message":[],"_revision_number":17},{"id":"4d7fecf8efe66415908f0e69a2cea7b29ab6917e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 08:41:44.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7549205500d34bb587d684224bbc3269 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/681c40804bdf427eb419075bf4745b35 : SUCCESS in 9m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9630ee820a2423aa1cc4fa58b0a08c0 : SUCCESS in 9m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69c446e753854ef18700b22bc9b47aa5 : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eea096378854195a454e3a0c6ef6e5e : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/796206f9781e47f3bddbf5cb264a3f1b : SUCCESS in 10m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5c53ea1ad92480cbd1eaf00e9adfe6e : SUCCESS in 6m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2430c94478b4ad08f8f3ddda0bbeb4d : FAILURE in 12m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/52c7101d0b4149ea93d428457d616402 : FAILURE in 12m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/672dcabb4c404fef9b42cfac2e34babe : SUCCESS in 8m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ce42c4453424339b49e0242fe3f77ce : SUCCESS in 13m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76ec71c96f3c48ea84c5417e16347999 : SUCCESS in 10m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5c382083b6c54ffaa708828487199218 : SUCCESS in 25m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e16e490f2450498b8b6f43cd0a5d1b28 : FAILURE in 26m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/42058d6791c44dd1abe7ea296c819bb6 : FAILURE in 19m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2c3b5f8aca7446cea87ac8146189ffed : FAILURE in 17m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c5f47cdbb194ef8a500909a4fca8b47 : SUCCESS in 21m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/27dcd047cce24d4b99f0f83aa3273791 : SUCCESS in 7m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6dec9c160493411eb69501daeac920f0 : SUCCESS in 9m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b413172a7fd4f098984b233dab39c8f : SUCCESS in 11m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a8795c1db543484bbf1af49164eccd1d : FAILURE in 13m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c78b0e38738e4a8ebd4485520c345c2b : SUCCESS in 14m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef4fc059df034269b65688d327a3dcf2 : SUCCESS in 54m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbdfaf5996c7426c86ceed7881521465 : SUCCESS in 1h 11m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f6f74571a744def89c714d65defce05 : SUCCESS in 1h 16m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1a3dbf747d4d43ad8d419e5cfd589c52 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":17},{"id":"a420a5323a9f56555d5a522f6b4e972dbe2f87cd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-30 06:18:37.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"7e31c34eba3a320127a30fbbec01ea9d7960f441","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-30 06:19:00.000000000","message":"Patch Set 18:\n\nUnable to freeze job graph: The nodeset \"ubuntu-bionic\" was not found.","accounts_in_message":[],"_revision_number":18},{"id":"d5e3c988e05ce9f193a97d1014d6beaa830495a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:20:51.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac01dc969984d009f90e4985fa79aba : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a162448a684e4b8364f6f5d06a4903 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c527a14fd480474c835baa37faf2efb6 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07bedbf225cd4601ac80a87792040510 : SUCCESS in 7m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/828c993764f14656a5f13cf58d272460 : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9a1278853f84b5688d9b701b6196db6 : SUCCESS in 5m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/82e466044b2f4745a1cc211f549b2aaa : FAILURE in 15m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7ad96efe6ee343d68813155e82bec72e : FAILURE in 16m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/34a4f573960743b5bf703fc786350c9c : SUCCESS in 8m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/952a54823ca846bc85d9efe930e663de : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c167da472da422c85d3a3a812599f7b : SUCCESS in 9m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d08056eb72114ea3bf0701bf58eed744 : SUCCESS in 15m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/845c5573680d4e62abe14ced1d213e63 : FAILURE in 28m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e1787b405478436fb0b4fe70ae9cb2b7 : FAILURE in 16m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/86da599c0c7e4409950cc644e3e6d9a9 : FAILURE in 16m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cff7cbdd5581409c8518b0731cf13f84 : SUCCESS in 22m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0ff86fc99fa54af7851287e1ce2b40f4 : SUCCESS in 7m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b2b9cb3428614c3c911df010fe31597e : SUCCESS in 7m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c385932f70c146d1b0932eb6a001af5d : SUCCESS in 7m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8b4575bf90ed42439edf51309b48368a : SUCCESS in 10m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/48c758ab9929404bb09ce944ca5f0676 : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91551ea856c448b4be121030464c6b23 : SUCCESS in 54m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5df45fca798140c298a1b4fabb8a854d : FAILURE in 1h 14m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fa8081f76f84603882462b929d00a74 : SUCCESS in 1h 26m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/72c2725aaaeb495785255a2dba9c54f0 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":18},{"id":"1ce2ebbff44c7c276bccc98702c0e5a237290288","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-01 06:23:48.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"55608a7ea91319bb33da89a797a4a4db269b710c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 07:51:13.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6024334d342b4ced8473ff123b9ed094 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dfb17266d594c18b6156e5c9ce2605b : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1554b48529946d29bd51b123041844a : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e70e86e996a14d4d9eb1ddb25de82b57 : SUCCESS in 7m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aef79f3bed1c45828ebacbcc12234b7c : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d90f6546c15e45a1b1899ba0f507b2bb : SUCCESS in 6m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9da6e36622ee48ce8c56c791482f0693 : FAILURE in 12m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cbbeb5e62b19418781192a0bf3419912 : FAILURE in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e0cf769a6c3e481bacdf8fd41acd6a74 : SUCCESS in 6m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d9e29ff5a9694b79971ae78647c0c629 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0082dc4125114737b6492e5e44af84da : SUCCESS in 8m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5f63ada6a6334d03b7f3d0c86cb463ec : SUCCESS in 16m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4744bf4fe3f746c6b417865709b31539 : SUCCESS in 28m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/428162ff9f6a4c92881e224d54247581 : FAILURE in 29m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d76bd0d5afbf45c59e11557cf72168e0 : FAILURE in 16m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49d3ecdb7a954653983319a2ae2a7bc6 : SUCCESS in 16m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/245bb2cc6d9a43b4984a45e999e52a5a : SUCCESS in 5m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/811f6e15121d4e2fb8e8e48d376a1201 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1bf0768149684e1bac3a8552be1bbc0b : SUCCESS in 8m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/890b7522ec7e4cbbb5f9af74c3720ec2 : SUCCESS in 10m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f18f7acc5fc34f979547903561ea0c34 : SUCCESS in 18m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c993696750414677ad57d609208680bd : SUCCESS in 1h 09m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/90a50b2178714fd58f490d6921f54ec5 : FAILURE in 1h 18m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4f13a50ba7a4a5fa8b7c100ed29fba7 : SUCCESS in 1h 20m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cb6c9deb37d94dd19ec4324a9f30fc4c : SUCCESS in 1h 08m 30s","accounts_in_message":[],"_revision_number":19},{"id":"624c5777ac79be2cb74e9d4bd5eaa88f377904ff","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-02 06:59:36.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"da72c7342b327df21b3110c4cbedd0329918d1e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 08:19:37.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df0c6dff02f94344a514a8bdefba4ba3 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a09a1d8a5044db396207eef888925b4 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b497190a50644876a8e2f36044e2e270 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b42aa8a0724846629aa3e23c1beaf27e : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56885f3aa8f94d8b8c479f127101df94 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d729bbea66664aa0a7543b039f8ed68e : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f16a230364994717a777a10d915373f7 : FAILURE in 7m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/69eccd0f71e041159ba44f418e16fdcc : FAILURE in 8m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/27bc13b8e30948ce8875566988573c3b : SUCCESS in 4m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b74cc0111e884167afc41af0c2eb1530 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6694909f14954731bd4e7d360ccd6372 : SUCCESS in 7m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c4904fa1408f40c781e18fea81bddade : SUCCESS in 12m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cb931746c4bd4841a782fa1e70a9ad51 : SUCCESS in 28m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/beb83134e63f4569b6df16ae959d6243 : FAILURE in 14m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a97825fc83944b27b2b9709f1fc60dea : FAILURE in 14m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23bc6ca08bb64f05aecbfed0935cd7b7 : SUCCESS in 18m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a84e3557683b41c4802d1dfb3756466e : SUCCESS in 4m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cd8423de8c0246b898d19f2d46f1b5be : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5dbbf2e34544393beceb651d67de179 : SUCCESS in 6m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8981b309ae3748adbfeeb80d56a29ba1 : SUCCESS in 9m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dbfea60ae65649e5bf8fe01c6be71f3d : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8b0f733badf47a5a3b5d9d86e69b896 : SUCCESS in 51m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e7ec87394df49448c8cafe4b9d88a80 : FAILURE in 1h 07m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b61aa75a446c47c4919e011935ca9a4f : SUCCESS in 1h 14m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fc7bffd47a77409b88073f5ad9cb7ceb : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":20},{"id":"63d3e30134723bed3a520deb355d743588ba6654","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 18:22:16.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"903bafd6763e1a39a602c0c1bfb5f7c05b647d1c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 18:22:25.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":21},{"id":"f0a821ecbcd38151df189d8ce4746c28615a8d4b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 18:35:27.000000000","message":"Patch Set 21: -Code-Review","accounts_in_message":[],"_revision_number":21},{"id":"b7ec1b8653cfe975fda33418d67fe16fd6e30d6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 20:02:05.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19cdb37d60547bd8e3e7e4470f936a6 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9c845cdf4942bf917424d796582251 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dd752c12da54681a77659c2be914abc : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2506d3bd17f847ef9b1d510c41bb50f1 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5500fa5b50fe40928b3a11d11ac39ace : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93955edac64745e8873e2d42ef81305f : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f56db2ca04a143b2a7445e3ad47100c7 : FAILURE in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c6145b91cd484e109aac157bfa04b378 : FAILURE in 8m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4cefc9415ae4d8b8ae012d144e0b5a7 : SUCCESS in 4m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8fda094f26cd48f39367b244ee10e96e : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1bedcc9acd3f4a09ac015913e76f9623 : SUCCESS in 6m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4a4d17909be1470e85b46104ee6b9e13 : SUCCESS in 13m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5c1b9cc612bf42d19baef12052d67592 : SUCCESS in 26m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6d1af44886554ed498835a506d4b4c65 : SUCCESS in 12m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9899cc71b5fd41e98ced2df06a05b29f : SUCCESS in 12m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f9def1a153e404392cff182bda2ba19 : FAILURE in 17m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fcd9e552bddc475ca939e1851547d6ed : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9e8dcef0f2074b9b9ba11492a4a20a11 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c33699e873c84be190e34cc7072a7fdb : SUCCESS in 5m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/157d907460a44e7e8d1fa741af6a7c00 : SUCCESS in 9m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/83676fa44fd647e4a7ab1e41ee4baac5 : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b8eda338625484b9385847a40df9bd4 : SUCCESS in 1h 07m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6806080b17544a9a3b2060f0617488b : SUCCESS in 1h 02m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91e1f9d4841c41b5bf851700b662b862 : SUCCESS in 1h 38m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/47399d9970a54ec98ea1127a987a5781 : SUCCESS in 55m 29s","accounts_in_message":[],"_revision_number":21},{"id":"15b21a8150900a41793aafc57fe06a75350ffeea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 20:43:02.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"8269be1235da7c40d9df712f161488d686632ab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 22:10:26.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e597767bb0474ec694177ccf7a4a1970 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b3527b5034748e3960742f9715a38d1 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f610ff0ab3b745e78979e76993efa689 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52249577c4cc4ee083ec350c1bf8dd7a : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70f4833b493f4a39b08fb60190557ea8 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ed49f19676647f29cddfe865800fef1 : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91cd1cbb7cca468989074340177f7a9c : SUCCESS in 7m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cd198be554ec4f7e97e2bd12e5739e93 : FAILURE in 8m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6810ffcaf685474a8d7de34dedd300da : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7c25a4d1b6eb409ba8d2948b7af7f80a : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de2db488d0a9418e86a9eebb30747e70 : SUCCESS in 8m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4094cbb45c1047bb89b9498573b63447 : SUCCESS in 14m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d1cdd0e370ea4f348938b7f456967986 : SUCCESS in 28m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9cb29ea51dd040e1930483c66d9a094e : SUCCESS in 13m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/efe4f537ed854b499a8411778e0ba541 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e9416ab36cd4f4d9ae425f4f5234162 : SUCCESS in 15m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9755a7c2e67549f3ac334619b77f66d9 : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3475199938a3467a9a89c1cd74af3ab4 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43c51ec2ab0d4259916e63546879bf25 : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1edc23d784974a02874f75cb933ac88c : SUCCESS in 12m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d762d2d49ecc4931a5f324beef24d98d : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/469e5168b1a24fd38bfdcc56b279d7eb : SUCCESS in 51m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4479118916e7456193dba4b82bdd94fb : SUCCESS in 1h 00m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6990638e7ee0452994f7e43758558be2 : SUCCESS in 1h 22m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/299572d90d3b44ab9d7cdfc133ac4f5e : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":22},{"id":"9e2d2fb32becf4990127ac7aec50109f9b340dfd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 22:24:48.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"8bc5216865babf415d14e3f9ac5b14b1cbf8f797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 23:47:15.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131a11d74f4e48fba90a0fdedad944b5 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/265b16953da64c2d92955dce5c3ba301 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5195ea7e2ccf4be2ace4ee0fb59a2919 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe25415797a64936969252a2653b84ce : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68fe8a0bbfd24c2caca205fe6f94aa12 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb6f73870cf3411c80c313c3f76b7e34 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1739987c4146436b9dcbb33a308cb0d5 : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a9f3f1ac6834e139b4694ece90cc11d : SUCCESS in 14m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7fbe37dae5f74ab9805988c4dea4c961 : SUCCESS in 5m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ce539be710914730895342429dfa07e8 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1cd0b89570f418e960fe7dfd04aff32 : SUCCESS in 8m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/40a21606f2c24425ae4d94fd8a4abb47 : SUCCESS in 18m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce67a502ad70419a8e6055c8abbe5766 : SUCCESS in 22m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e0b18e8be85f4c40b44e5645a7df8f28 : SUCCESS in 11m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4422f9e710b4436f87e79f92aef3d4c9 : SUCCESS in 13m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/802ef0595bf3481cb9d465b9c6d91acc : SUCCESS in 16m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4bbba725d44e4cbdafb8c990a0d6f34c : SUCCESS in 4m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9465e93cf74941589576634b0843e850 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0adc2962568a4d9280d6772ba9fb44c7 : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b3ba7c21a3b64c5f94d3b1c49cb20001 : SUCCESS in 12m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b7cfeec41c5a44eaa48c71f9d3572be7 : SUCCESS in 9m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f81516d987f145ef96eab1ea41beec7c : SUCCESS in 53m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/64569abc24864ec4b502f65dea7e26d9 : SUCCESS in 1h 02m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c326d70202b44cbb6dbe670396cad9e : SUCCESS in 1h 17m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f109ceab236846378b8ca2825f216e92 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":23},{"id":"2bf443ae3e56ce615018ad63175e817c4d58ff78","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 23:59:19.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"3d1bd997b792f65a6facf9ff3fd0819651e12818","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-03 03:08:51.000000000","message":"Patch Set 23: Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"2f1e5415f7fd7ca281c71d0954ed5d96d17907df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 03:09:03.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"ef9b5dc97ff9677e5bd5b4da6ff5d98963e5f44d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 04:26:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":23},{"id":"5d99ce7b7d5066985ac3e294fa25840c41ed0b46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 04:26:25.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e50c70352846a698e0b7b11d23b86d : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f1aa330f134fb4aa7cc10c8fedeb3f : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e23020cce1dd434a815fc11843230900 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbe12221146b4f3fb537b222994f20ad : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e944721d136447fd9db6718bdd93c85d : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ccd96285448644ada25a5df8c0881089 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/58a0c56de64c402fb8eda2433d7a22c4 : SUCCESS in 7m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/176752602fed43c3bddbf3cb57260568 : SUCCESS in 8m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/33a718912590442386ceb104583a97ed : SUCCESS in 5m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/831ba756e9e74fb794bd5f897e0906de : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67685481c16f4433ae90e6c496f5dde2 : SUCCESS in 6m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/87f8c6d832fa469fb4262e184a15a99e : SUCCESS in 19m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/16fdce232181458bb570858a03cb3b47 : SUCCESS in 21m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8c6006f7ea404f3e992f5d391366ede4 : SUCCESS in 12m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ec8f2dcce3c04c5fb27e4c1c90521410 : SUCCESS in 13m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97b5ae642bfa41db9a5159235ef1e961 : SUCCESS in 15m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ffa25e582d5b4a1f91832cf4fdd11d2d : SUCCESS in 5m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/374c8536f56045f2a91a6934889b410a : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56e3603bc7a945448caf5b5a0a46ba63 : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3af2a2b38ff34026b4eba76e3197e494 : SUCCESS in 10m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ecd526f52c8743a1a3ae72b73f93df8e : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5becab65d1194bfa980f607864157ab9 : SUCCESS in 51m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7dcb7766a7f42939b1de93aca08bb8f : SUCCESS in 57m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f636734fbd64e108f77dbe0de43e32c : SUCCESS in 1h 13m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b824e1348b64c2a829f28dbb0e22d8a : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":23},{"id":"0d0718faa289cbd3aad68e7bc2610bea3e705fd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 04:27:36.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/176ad8e6c3aa42b1b41ab9683b54160b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"b21c994c07c9b43b756a46dfe2a1577874a09d79","revisions":{"9fcbf75d673be9f4995ee2f120b0f1da8bb8aeb0":{"kind":"REWORK","_number":1,"created":"2020-04-12 16:22:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/1"}}},"commit":{"parents":[{"commit":"5dff9df1334af77e9ef2d539ef838cce6e8ea1c2","subject":"Merge \"update constraint for python-zunclient to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5dff9df1334af77e9ef2d539ef838cce6e8ea1c2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-12 16:22:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-12 16:22:22.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fcbf75d673be9f4995ee2f120b0f1da8bb8aeb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fcbf75d673be9f4995ee2f120b0f1da8bb8aeb0"}]},"branch":"refs/heads/master"},"685499a00a7899d2aa146603abb3cf16375db97f":{"kind":"REWORK","_number":2,"created":"2020-04-13 06:13:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/2"}}},"commit":{"parents":[{"commit":"d10b75e5cc0c09928ff340f3b1919a099400de9f","subject":"Prepare upper-constraints for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d10b75e5cc0c09928ff340f3b1919a099400de9f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-13 06:13:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-13 06:13:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/685499a00a7899d2aa146603abb3cf16375db97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/685499a00a7899d2aa146603abb3cf16375db97f"}]},"branch":"refs/heads/master"},"39d3af65f2989a044488961d68689967d8b39247":{"kind":"REWORK","_number":3,"created":"2020-04-14 06:16:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/3"}}},"commit":{"parents":[{"commit":"31371721fb5c8c038690cbd9a2629a1b546e6f23","subject":"update constraint for python-glanceclient to new release 3.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31371721fb5c8c038690cbd9a2629a1b546e6f23"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-14 06:16:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-14 06:16:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39d3af65f2989a044488961d68689967d8b39247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/39d3af65f2989a044488961d68689967d8b39247"}]},"branch":"refs/heads/master"},"45750502b9f32bad71ee3a1148cb803f8df9f08c":{"kind":"REWORK","_number":4,"created":"2020-04-15 06:20:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/4"}}},"commit":{"parents":[{"commit":"31371721fb5c8c038690cbd9a2629a1b546e6f23","subject":"update constraint for python-glanceclient to new release 3.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31371721fb5c8c038690cbd9a2629a1b546e6f23"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-15 06:20:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-15 06:20:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45750502b9f32bad71ee3a1148cb803f8df9f08c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45750502b9f32bad71ee3a1148cb803f8df9f08c"}]},"branch":"refs/heads/master"},"6cca924c04c02d8bb6d8e5a13fe7a7bff5ad3e20":{"kind":"REWORK","_number":5,"created":"2020-04-17 06:16:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/5"}}},"commit":{"parents":[{"commit":"2455e78f0de675d3ce896b48e7332dc429ccfc11","subject":"Update constraint for castellan for 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2455e78f0de675d3ce896b48e7332dc429ccfc11"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-17 06:16:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-17 06:16:41.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cca924c04c02d8bb6d8e5a13fe7a7bff5ad3e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6cca924c04c02d8bb6d8e5a13fe7a7bff5ad3e20"}]},"branch":"refs/heads/master"},"76ebec16302f69ea37f5298fa7a4dbf0df9ad7ab":{"kind":"REWORK","_number":6,"created":"2020-04-18 06:21:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/6"}}},"commit":{"parents":[{"commit":"2455e78f0de675d3ce896b48e7332dc429ccfc11","subject":"Update constraint for castellan for 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2455e78f0de675d3ce896b48e7332dc429ccfc11"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-18 06:21:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-18 06:21:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76ebec16302f69ea37f5298fa7a4dbf0df9ad7ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76ebec16302f69ea37f5298fa7a4dbf0df9ad7ab"}]},"branch":"refs/heads/master"},"a1b276d46275e66bcab9d7df7115388631171a4d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-04-19 06:22:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/7"}}},"commit":{"parents":[{"commit":"f55297cc5737c5f5b866431ad9e9e42116039582","subject":"update constraint for python-monascaclient to new release 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f55297cc5737c5f5b866431ad9e9e42116039582"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-19 06:22:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-19 06:22:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1b276d46275e66bcab9d7df7115388631171a4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1b276d46275e66bcab9d7df7115388631171a4d"}]},"branch":"refs/heads/master"},"15c1c9e5e69a5e4040fdd6eb040bb4f4f94321f1":{"kind":"REWORK","_number":8,"created":"2020-04-20 06:19:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/8"}}},"commit":{"parents":[{"commit":"f55297cc5737c5f5b866431ad9e9e42116039582","subject":"update constraint for python-monascaclient to new release 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f55297cc5737c5f5b866431ad9e9e42116039582"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-20 06:19:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-20 06:19:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15c1c9e5e69a5e4040fdd6eb040bb4f4f94321f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15c1c9e5e69a5e4040fdd6eb040bb4f4f94321f1"}]},"branch":"refs/heads/master"},"3b12900c1c5f3ae38489006c80000553839fec97":{"kind":"REWORK","_number":9,"created":"2020-04-21 06:15:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/9"}}},"commit":{"parents":[{"commit":"f55297cc5737c5f5b866431ad9e9e42116039582","subject":"update constraint for python-monascaclient to new release 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f55297cc5737c5f5b866431ad9e9e42116039582"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-21 06:15:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-21 06:15:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b12900c1c5f3ae38489006c80000553839fec97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b12900c1c5f3ae38489006c80000553839fec97"}]},"branch":"refs/heads/master"},"de4d70e17438b1d1e111bb53d129e858b9b88f88":{"kind":"REWORK","_number":10,"created":"2020-04-22 06:13:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/10"}}},"commit":{"parents":[{"commit":"f052a629a5c626a929dbec50d99ad7e393e5bdff","subject":"Block confluent-kafka\u003d\u003d1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f052a629a5c626a929dbec50d99ad7e393e5bdff"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-22 06:13:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-22 06:13:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de4d70e17438b1d1e111bb53d129e858b9b88f88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de4d70e17438b1d1e111bb53d129e858b9b88f88"}]},"branch":"refs/heads/master"},"df7941c9066fc7b3639404617c318820a4ac9a9d":{"kind":"REWORK","_number":11,"created":"2020-04-23 06:14:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/11"}}},"commit":{"parents":[{"commit":"cd6b6f36247109a4d80853ec4cd42de8a8279971","subject":"Merge \"update constraint for ceilometer to new release 14.0.0.0rc1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd6b6f36247109a4d80853ec4cd42de8a8279971"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-23 06:14:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-23 06:14:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df7941c9066fc7b3639404617c318820a4ac9a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df7941c9066fc7b3639404617c318820a4ac9a9d"}]},"branch":"refs/heads/master"},"57fcc6fcf0c7c947c588d6fa2100c5ad70ebe27d":{"kind":"REWORK","_number":12,"created":"2020-04-24 06:13:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/12"}}},"commit":{"parents":[{"commit":"bc74822f08abd3598ff95f7703576a583eff7780","subject":"update constraint for tempest to new release 24.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc74822f08abd3598ff95f7703576a583eff7780"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-24 06:13:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-24 06:13:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57fcc6fcf0c7c947c588d6fa2100c5ad70ebe27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57fcc6fcf0c7c947c588d6fa2100c5ad70ebe27d"}]},"branch":"refs/heads/master"},"439b3a4b15165c93dbe5cbe7a6a05c6fb56f28ed":{"kind":"REWORK","_number":13,"created":"2020-04-25 06:45:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/13"}}},"commit":{"parents":[{"commit":"3b425cecd6d5238203e1903fb0df40c127a4e59e","subject":"Merge \"add contributing docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b425cecd6d5238203e1903fb0df40c127a4e59e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-25 06:45:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-25 06:45:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/439b3a4b15165c93dbe5cbe7a6a05c6fb56f28ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/439b3a4b15165c93dbe5cbe7a6a05c6fb56f28ed"}]},"branch":"refs/heads/master"},"3e02494ce99f31eaebec22ebf26e4647830624ef":{"kind":"REWORK","_number":14,"created":"2020-04-26 06:13:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/14"}}},"commit":{"parents":[{"commit":"3b425cecd6d5238203e1903fb0df40c127a4e59e","subject":"Merge \"add contributing docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b425cecd6d5238203e1903fb0df40c127a4e59e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-26 06:13:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-26 06:13:20.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e02494ce99f31eaebec22ebf26e4647830624ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e02494ce99f31eaebec22ebf26e4647830624ef"}]},"branch":"refs/heads/master"},"07d1afa2ad98bbd95290ec158afdeddad1c1b1b6":{"kind":"REWORK","_number":15,"created":"2020-04-27 06:59:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/15"}}},"commit":{"parents":[{"commit":"3b425cecd6d5238203e1903fb0df40c127a4e59e","subject":"Merge \"add contributing docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b425cecd6d5238203e1903fb0df40c127a4e59e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-27 06:59:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-27 06:59:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07d1afa2ad98bbd95290ec158afdeddad1c1b1b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07d1afa2ad98bbd95290ec158afdeddad1c1b1b6"}]},"branch":"refs/heads/master"},"bf59011d23b780a1f4a07faf18dc841ae56d2736":{"kind":"REWORK","_number":16,"created":"2020-04-28 06:15:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/16"}}},"commit":{"parents":[{"commit":"3b425cecd6d5238203e1903fb0df40c127a4e59e","subject":"Merge \"add contributing docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b425cecd6d5238203e1903fb0df40c127a4e59e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-28 06:15:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-28 06:15:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf59011d23b780a1f4a07faf18dc841ae56d2736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bf59011d23b780a1f4a07faf18dc841ae56d2736"}]},"branch":"refs/heads/master"},"ffdf7a2f88f294b4e98b6d7ad27ed4c900871f7f":{"kind":"REWORK","_number":17,"created":"2020-04-29 06:22:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/17"}}},"commit":{"parents":[{"commit":"b85e113e6c2388482ffc8b778c8757ba49ad4615","subject":"Merge \"Add validations-libs and validations-common to openstack_requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b85e113e6c2388482ffc8b778c8757ba49ad4615"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-29 06:22:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-29 06:22:07.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffdf7a2f88f294b4e98b6d7ad27ed4c900871f7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffdf7a2f88f294b4e98b6d7ad27ed4c900871f7f"}]},"branch":"refs/heads/master"},"21991450e2be0dcd9c3439ab94cf03ea53573d76":{"kind":"REWORK","_number":18,"created":"2020-04-30 06:18:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/18"}}},"commit":{"parents":[{"commit":"c4223273661b6dbe3e59437061322aa3eff387ee","subject":"Merge \"update constraint for openstack-release-test to new release 3.0.0.0rc1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4223273661b6dbe3e59437061322aa3eff387ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-30 06:18:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-30 06:18:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21991450e2be0dcd9c3439ab94cf03ea53573d76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21991450e2be0dcd9c3439ab94cf03ea53573d76"}]},"branch":"refs/heads/master"},"e758a9d9b46a3923b3d34879fc2d8af74dc741c9":{"kind":"REWORK","_number":19,"created":"2020-05-01 06:23:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/19"}}},"commit":{"parents":[{"commit":"cf54480ba9ad4ef0032b861ee3f8d091823eb3fa","subject":"Remove limitation on python construct library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf54480ba9ad4ef0032b861ee3f8d091823eb3fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-01 06:23:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-01 06:23:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e758a9d9b46a3923b3d34879fc2d8af74dc741c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e758a9d9b46a3923b3d34879fc2d8af74dc741c9"}]},"branch":"refs/heads/master"},"d83c0671f86256ce6aa6724852f435c774d08bd6":{"kind":"REWORK","_number":20,"created":"2020-05-02 06:59:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/719345/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/20"}}},"commit":{"parents":[{"commit":"75c50a27b0c13c9befc8f8324a31fd08db34df4a","subject":"Remove unused packages from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75c50a27b0c13c9befc8f8324a31fd08db34df4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-02 06:59:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-02 06:59:35.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d83c0671f86256ce6aa6724852f435c774d08bd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d83c0671f86256ce6aa6724852f435c774d08bd6"}]},"branch":"refs/heads/master"},"8b59a961e64cba3bcb9ef8dcc14208b41d644732":{"kind":"REWORK","_number":21,"created":"2020-05-02 18:22:16.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/719345/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/21"}}},"commit":{"parents":[{"commit":"75c50a27b0c13c9befc8f8324a31fd08db34df4a","subject":"Remove unused packages from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75c50a27b0c13c9befc8f8324a31fd08db34df4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-02 06:59:35.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-02 18:22:08.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b59a961e64cba3bcb9ef8dcc14208b41d644732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b59a961e64cba3bcb9ef8dcc14208b41d644732"}]},"branch":"refs/heads/master"},"f6de6c25de572e7781024876d0ff751a695c9975":{"kind":"REWORK","_number":22,"created":"2020-05-02 20:43:02.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/719345/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/22"}}},"commit":{"parents":[{"commit":"75c50a27b0c13c9befc8f8324a31fd08db34df4a","subject":"Remove unused packages from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75c50a27b0c13c9befc8f8324a31fd08db34df4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-02 06:59:35.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-02 20:42:53.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6de6c25de572e7781024876d0ff751a695c9975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6de6c25de572e7781024876d0ff751a695c9975"}]},"branch":"refs/heads/master"},"b21c994c07c9b43b756a46dfe2a1577874a09d79":{"kind":"REWORK","_number":23,"created":"2020-05-02 22:24:48.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/45/719345/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/719345/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/719345/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/719345/23"}}},"commit":{"parents":[{"commit":"75c50a27b0c13c9befc8f8324a31fd08db34df4a","subject":"Remove unused packages from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75c50a27b0c13c9befc8f8324a31fd08db34df4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-02 06:59:35.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-02 22:24:36.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieaab5d43b6172ca9fc88d842e8e76e9c44db1b0b\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b21c994c07c9b43b756a46dfe2a1577874a09d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b21c994c07c9b43b756a46dfe2a1577874a09d79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
