)]}'
{"id":"openstack%2Frequirements~719611","triplet_id":"openstack%2Frequirements~master~Ieb159dd65a9044ec74baa20370a3b00b0f755cd1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ieb159dd65a9044ec74baa20370a3b00b0f755cd1","subject":"update constraint for castellan to new release 3.0.1","status":"MERGED","created":"2020-04-13 17:22:03.000000000","updated":"2020-04-13 21:04:02.000000000","submitted":"2020-04-13 21:02:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719611-1586811750316-b16e9fe7","meta_rev_id":"0ea90919b74f283ff8148f9849e4a3c1fbc580f4","_number":719611,"virtual_id_number":719611,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 21:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-13 17:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-13 19:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-13 19:33:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 17:29:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-13 19:33:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-13 21:02:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecc7c1052e23e1ee9ab39073dd80c7f041dc0b6e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-13 17:22:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c28d09a2e5e8608fea3e43fd2a45bbde0dfa5d1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-13 17:29:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c301929a7e18081f050e91a6b3a93e71923b834a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 18:56:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3e8e72b0c84c40babedf1bf2bb873d : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911e2cca72114a808231b158bf9f557a : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb454e6de7b74e37b8cd3f447beceb3a : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88a8aa79f3f34791859f399488bdac7d : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d6d856b399641e882d15eec9fa6ef47 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/918d925eec244c85873b86f70b1b1f7b : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4143c50e067464f96ad93135135f983 : SUCCESS in 5m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c6e0cbd901494a439d29e29d929b21f8 : SUCCESS in 12m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/40f7b704de52480e8b9d30c2b5f1911a : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d6ddce855f794903813fee9f27d3e423 : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/41635921a1864590a1b687b213f1ba5a : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/204e496255994dfeb1bec1996615e5d6 : SUCCESS in 9m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/52957c2e984d492d98b173713738e29b : SUCCESS in 21m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7d69deea6a64bc2a2456bc0bc2ed7cd : SUCCESS in 30m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fefc93e5f62f467396b28fc4c49afa1f : SUCCESS in 18m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e9fc474cc132436faaa9b285ed41ef77 : SUCCESS in 18m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be761b3e10de4b8484a7bb8784d1f735 : SUCCESS in 15m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4ec7ec29829a4d4b8baebc6229d9115d : SUCCESS in 3m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3aa0bae388594066b1b020783a83af05 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72a08e06b8f84e56915cf7ed2a971789 : SUCCESS in 8m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bc280f0046414c36a2cfff3be948a56b : SUCCESS in 9m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/72527d8ce9a942b1b77dfedac0c68243 : SUCCESS in 21m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b384cafa9bfe4361bbacfc6b433586f6 : SUCCESS in 53m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f88b7035e51b479e81fe2aea1ebe2823 : SUCCESS in 1h 20m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b7cc2d0cc574817b151a759aa041c71 : SUCCESS in 1h 25m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a6c9f7309684997a85dd0855e86ae26 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ab279cdad61a107db056392446b4c269c8b61cb5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-13 19:33:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8a2c530bbd1b9f2be521db3b5f9edd9df30f71a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 19:33:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1cf70cae74a0ec8016dfee45e1b1e70d54032a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 21:02:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e4e62903c14c2f91bdcb6b633798de : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2926cc16c189463cad4f90ea9fff12da : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7762ebb805aa473484cfd3438c080811 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/069f52ddd753430f9de266a77cf96401 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1be7321198d54d3b9d778670e8f4eb48 : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38c3b1022249466ba9816826f7d9e82f : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb8d869139ae4c618eb5f0dd547582b9 : SUCCESS in 8m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4d505e2c5be44f169970e9f0b8af9e63 : SUCCESS in 10m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e7cee4f0778a41ceb7539b1416ee30e6 : SUCCESS in 4m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/556d432244fe47759d39974cc526383b : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b2fdf76a507457e859d34e9d52d8622 : SUCCESS in 7m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/722d4fb3d659481c9d18e8949f846d9a : SUCCESS in 14m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec557bd2adcb4437affe139ec61e4cf4 : SUCCESS in 34m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5a751c31bd39407787f5d7fd0e29d051 : SUCCESS in 18m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/057782b1bdbe49b9bf2be2b90c8fff64 : SUCCESS in 19m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d288b9cdc7634570899fa3e7e9359e57 : SUCCESS in 23m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/025fb677f1ba4fdc8842e0690ed9f179 : SUCCESS in 4m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/91461177b9b84e359a132f4ce28a9c61 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b76d9a66a63f49c58388018c537adc70 : SUCCESS in 8m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/105d0494977b45e19f04500c03ffd501 : SUCCESS in 13m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec1babc7f0fa48a7a700bc27d7ca6833 : SUCCESS in 13m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3083bf12161d42d5ab56f674b1a06330 : SUCCESS in 50m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/835e45c5f58b4a33861241dab27d55ce : SUCCESS in 1h 13m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8e3a549e13742f9849fd0ff482d671b : SUCCESS in 1h 11m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/befe44b7e4f6408db6d7d62fc562a854 : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":1},{"id":"40d1a39611eec7a093a5ebbe032f7fc01b19a771","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 21:02:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5a1f0562fd9ec54223108113a0cdc60b0ad00809","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 21:04:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7caf115472b4bae92d93e51f93f9595 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"148547409985e23a93cd51b9923856c4f2146173","revisions":{"148547409985e23a93cd51b9923856c4f2146173":{"kind":"REWORK","_number":1,"created":"2020-04-13 17:22:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/719611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/719611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/719611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/719611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/719611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/719611/1"}}},"commit":{"parents":[{"commit":"1af40a1b316b85f9ee12dc6850031e8994e427d1","subject":"Merge \"Small docs cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1af40a1b316b85f9ee12dc6850031e8994e427d1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-13 17:22:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-13 17:22:02.000000000","tz":0},"subject":"update constraint for castellan to new release 3.0.1","message":"update constraint for castellan to new release 3.0.1\n\nChange-Id: Ieb159dd65a9044ec74baa20370a3b00b0f755cd1\nmeta:version: 3.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I7e0feec33fa9ecf5a975b2a1776f4ff1100ec2eb\nmeta:release:Code-Review+1: Douglas Mendizábal \u003cdmendiza@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/148547409985e23a93cd51b9923856c4f2146173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/148547409985e23a93cd51b9923856c4f2146173"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
