)]}'
{"id":"openstack%2Frequirements~720501","triplet_id":"openstack%2Frequirements~master~Id649ab6c871bb24042379539c1fe21371e35bbeb","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Id649ab6c871bb24042379539c1fe21371e35bbeb","subject":"Update constraint for castellan for 3.8","status":"MERGED","created":"2020-04-16 15:11:18.000000000","updated":"2020-04-16 21:43:33.000000000","submitted":"2020-04-16 21:41:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720501-1587073304860-332a2ac9","meta_rev_id":"3ac4a2b83344f23ab534a8ce9dc2f0fb7b04e329","_number":720501,"virtual_id_number":720501,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 21:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-16 15:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-16 15:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-16 15:47:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 15:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-16 15:47:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-16 21:41:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f20753ce3e8738062281d9e44f6366a7637e10a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-04-16 15:11:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cd898e6e6934df4a73d4c8251387993dafac070","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-16 15:20:07.000000000","message":"Patch Set 1: Code-Review+2\n\nShould be fine. Having any 3.8 entries here are really a stopgap until we 3.8 becomes official in Victoria. It will be important to make sure 3.8 gets added to the project\u0027s metadata once full testing is in place:\n\nhttps://opendev.org/openstack/castellan/src/branch/master/setup.cfg#L19","accounts_in_message":[],"_revision_number":1},{"id":"995cb40e93aff3b21a2f5f37a77f881dc6866f90","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-16 15:47:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c2bbdea03852d5490019f69fd956eaf0b426bb1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 20:07:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcddb14cea864dfa868b44fb52b56aa2 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506515f7e6ab4912aa483ef0aca28c6a : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c442682a82e1430a8fe75b9538b7de47 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55da490357a449f3ab3075c0fbc6b874 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c453116d44f1457b98ef328a318a454c : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36aed61cb0c7435383511ee3f21fb4c4 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6885c9cafa19419e94e5d94884f3fcee : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c78ed0f6d8ef482d971fe0f1c0ee1300 : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4487193049ef428182c960f60244e66a : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/897d8b3aa5494d05b51d363b6623d8bd : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/70e4b22ada2c4e178e693445c7e48f60 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26ab60f079af45d7af09de466b07467c : SUCCESS in 8m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/006d857c4a0d442083ddd4fc44a65482 : SUCCESS in 39m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8d2a8afbca0842a4bd21d3189b3dbdc9 : SUCCESS in 33m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/84e0a8e7b6b7427fa9840a8b69132bcb : SUCCESS in 12m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f363011401a34f1aa6f48e959267178b : SUCCESS in 13m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1e2705b662948a894c62cc4ce282a69 : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2cc70bd1f5a94590936b70b4e52cfb22 : SUCCESS in 6m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f61ce9b474af4506bdf3e9636216eff9 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/261b1ff0830943d39bd349fd96b43822 : SUCCESS in 6m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b830d55652c64b6098b570289a17dcb1 : SUCCESS in 10m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/62d1e05e53a541e4a130767039b4f436 : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92fe093837ee41a091c705cbe8b08fdb : SUCCESS in 56m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0687fa33dc5243bab1114081f264a592 : SUCCESS in 1h 37m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6938e678b1ad4c81a53e5d5728c1dd58 : SUCCESS in 1h 32m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a7c4843dacbe49f28506f45fc749335d : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":1},{"id":"164103311c4edecf51917247036e1b6369fd8b5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 20:07:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b99151324fb66c34cb3e1c73f8add7d3d2910bda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 21:41:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6495aab2e5c4dfda1ea3298f0321024 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6d90f455974725a6fe7e71440af5c5 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24ff6afba25e42708262b265063771e8 : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db705e26dfca4314b9115401f58a14df : SUCCESS in 8m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da74222675d146edba52206b9d6fbbae : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e1d845daf71842f3a990ce74929164a3 : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dd92c8a2d61d423db5c64c3690f0d629 : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/75a128e1e4184b2e98c21fff98f641d2 : SUCCESS in 10m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/62aa091be26d45d6a1d51ec73bb0a926 : SUCCESS in 6m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/40e472ab1cff40c98d68c475febf86ff : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/927b59d19585424f9cf5e2e1b330e4dc : SUCCESS in 9m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/934fd27c2ac64f2fb4e3728180f22a9c : SUCCESS in 15m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3753cd4f4033470f849df0c6e059e511 : SUCCESS in 29m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/34c62e414a6f43428d7200c6937d38a8 : SUCCESS in 14m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/12b07cede4ef49eb8fa6c32c6d19252c : SUCCESS in 17m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2265045a34aa438c9b66799e6ae11a7f : SUCCESS in 17m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f253212b9ced47a48f7c45c8246ff8cf : SUCCESS in 6m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1ffa91d50866404b83ab5cb945faa26c : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb014af5f7b647338fc055084e079d21 : SUCCESS in 7m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/90c7d3f701b04d7c971a4dc8e6ac13d0 : SUCCESS in 12m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9de57c0c132b457d9e9bc2c778691ffa : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b301d2b3da1944849393181748b6d10e : SUCCESS in 54m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfa42d95466e46ecba12bbc0c7f9a282 : SUCCESS in 1h 23m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1256daeb9cf45ce8ef63862c90f0f0a : SUCCESS in 1h 27m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2b8e7882aa8940d79e1c133a5d43ee02 : SUCCESS in 1h 14m 19s","accounts_in_message":[],"_revision_number":1},{"id":"35526e5b4935a1d5a4c89284c0c50c71022868bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 21:41:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3ac4a2b83344f23ab534a8ce9dc2f0fb7b04e329","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 21:43:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/048244cfc8ca4fb798ff7ac74a96f749 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2455e78f0de675d3ce896b48e7332dc429ccfc11","revisions":{"2455e78f0de675d3ce896b48e7332dc429ccfc11":{"kind":"REWORK","_number":1,"created":"2020-04-16 15:11:18.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/01/720501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/720501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/720501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/720501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/720501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/720501/1"}}},"commit":{"parents":[{"commit":"31371721fb5c8c038690cbd9a2629a1b546e6f23","subject":"update constraint for python-glanceclient to new release 3.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31371721fb5c8c038690cbd9a2629a1b546e6f23"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-04-16 15:09:56.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-04-16 15:09:56.000000000","tz":-300},"subject":"Update constraint for castellan for 3.8","message":"Update constraint for castellan for 3.8\n\nThe Python 3.8 upper constraint was missed by a previous\nautomated patch. [1]  This patch updates the constraint\nfor 3.8 to match 3.6 and 3.7.\n\n[1] https://review.opendev.org/#/c/719611/\n\nChange-Id: Id649ab6c871bb24042379539c1fe21371e35bbeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2455e78f0de675d3ce896b48e7332dc429ccfc11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2455e78f0de675d3ce896b48e7332dc429ccfc11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
