)]}'
{"id":"openstack%2Frequirements~721643","triplet_id":"openstack%2Frequirements~master~I2d9a7214bd50579b1803af1d0b4cf06ca89d1c47","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2d9a7214bd50579b1803af1d0b4cf06ca89d1c47","subject":"update constraint for openstack-release-test to new release 3.0.0.0rc1","status":"MERGED","created":"2020-04-21 15:04:18.000000000","updated":"2020-04-29 22:27:39.000000000","submitted":"2020-04-29 22:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721643-1588199153052-bde3a662","meta_rev_id":"2dbce1c14dd422afc7e9e9c830ce81522ba810cb","_number":721643,"virtual_id_number":721643,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-04-29 20:33:04.000000000","_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 22:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-29 20:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-04-29 20:32:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 20:32:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-29 20:33:04.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-29 22:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc470802e1ef1b928406f18028cfec4d9a2ad25e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-21 15:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f0bbb94bab8aa4a07eac9b737559fd79167a976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 17:15:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f5857f2c114a3a9988e10a8f5931e0 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9424ab1f98934591b7f5ff46a95e62f1 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f56a8d301954a88b602fcb66c310b71 : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/158a1c3adb1f4e7d9b907d23dcd35850 : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0988c15ded2341b1b16304af4d896b6d : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ef5f04d0ef24eb78cead7e14697cece : SUCCESS in 6m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0eae72384ec64c5e9d3aac091adfc274 : SUCCESS in 5m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9d321360a8f44759eeaf506ad58c223 : SUCCESS in 9m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a80bcc8b6f6f435996f2ad038e61f920 : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/30a067a40fff48e6ab152375a75eec0c : SUCCESS in 9m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b874c676a38443aa8bc841134d9d369a : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a49ec54214604ee7978b199479ba679e : SUCCESS in 12m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5a7adc87312847d6889b342dade19f9e : SUCCESS in 14m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb68b7a008e849238ef29de063fab907 : SUCCESS in 24m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/595603e4546f42289751b074ba669355 : SUCCESS in 12m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/96ef7d471e174144b95e731b187bd1ac : SUCCESS in 16m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18e56f2d26b44a20bc630c2b53aa2e20 : SUCCESS in 21m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0e17f74c3d80405ba0353b9b854997ca : SUCCESS in 5m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3e9d37a5ced4489394565b24dfe95f16 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1d026c5d65a4fa791e0847c4b2c6aa6 : SUCCESS in 8m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f7808a6ef1ae482baf69fc57957c27f9 : SUCCESS in 11m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/221759c25a3349858ebbfbfdf73a0c7a : SUCCESS in 17m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b426e9f4f8c84192a855c6f380ca396e : SUCCESS in 1h 00m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/01610c076a6b482b8342ca033aecbd47 : FAILURE in 8m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0c25180132847f6af3652e10f42ee7d : SUCCESS in 1h 34m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d99fe70efa234c70a8d04fd55eeddaba : FAILURE in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"92de52b1600c65af3eb9ebce200e6ce204b642bd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-21 21:58:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"12f8f7b3715caca6bb337113dc8d2740f5054f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 00:12:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cedb833ab8e48488f9f96257f07ef1c : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be7729722eb4da2b34153b2ddf904cf : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9245148f6a8547d8a1e755df857e5227 : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/105ee5e0e52a4bfe9dbe1c88fe7591e4 : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5954f68ea48f486ba19a26ab65f56967 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0dd0bc054bda45b2b7544800b81c27aa : SUCCESS in 7m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0bcff3300c7497cbf628939ee7bcb8c : SUCCESS in 7m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/69767242d55842aaa9b94912d1716a7b : SUCCESS in 13m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/72412df92ff04406b4d574ce0fa61263 : SUCCESS in 23m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5e4b94ad853e43be9724ca46e9691050 : SUCCESS in 8m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9cb25b7edc048ffbd1fea7896a63f84 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cfed0d3a16cc4a3a84710a0233b891f0 : SUCCESS in 11m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/23f41c2f94164a9c9564f6fb0a9e4979 : SUCCESS in 44m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/076c51a33f3c485785013b50068fe0cf : SUCCESS in 32m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/469356e33def4a82bc43fec49f679db8 : SUCCESS in 22m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4594f7a7244f48da98993130e3fd1414 : SUCCESS in 18m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa6025006e2a4951a33660c9d8801273 : SUCCESS in 31m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4c25477a657649d19608009bc0f83f82 : SUCCESS in 6m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ee4e3071cea64effa1f118a9d1d76a27 : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/791a47116beb4b86a8952766c39e77a3 : SUCCESS in 9m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3a2879e34d924d5082d59e2bb985266a : SUCCESS in 24m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/32ac1c155bab40f9bbece56d32242001 : SUCCESS in 24m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d86c0ce52bf94933b1ff57e75f93a7c3 : SUCCESS in 1h 07m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ca718f9bf8b461ab6229ac1267d6b1d : SUCCESS in 1h 31m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/878b885a17fd461494dbee0d0487c218 : SUCCESS in 1h 36m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0bf770a2e9214b529a3cef5dc97199b2 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ed9525e89885a765f14c0cc10d8d737c68f7ab9c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-28 03:43:48.000000000","message":"Patch Set 1: Workflow-1\n\nno because of freeze?","accounts_in_message":[],"_revision_number":1},{"id":"7f20004904fd361f059bddeab4dd80cd1b289a24","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-29 20:32:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"730fd171519a6c0dac20768d571eb68e5e3ebfe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 20:33:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b1a956babc64d980a2c36d198cf4b343fc18d06","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-29 20:33:04.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"ubuntu-bionic\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"1813b90a60ed1eef1c415d1c99657df22268f263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 22:25:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52854ecc48b04ea1ae4b72d15010b565 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/519c5ed6b6df43028a638a82e2a90b9b : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/111c244d1b0744fd9d41cfd17fbbc8a8 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3afd8fe9f8741129cf46bb0be542eae : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/daeb584f971649459ca02a80cfd27b6b : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a838de6a656645d9ab489c210fe057b1 : SUCCESS in 6m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59ca732c080f4df08d58185421d1aa12 : SUCCESS in 8m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d43a7641afe47c5b3458adeb9cf13ff : SUCCESS in 11m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e749a2b7f93a44d8b4625654e9c2c473 : SUCCESS in 8m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63909439042f49a7b41573c4b3e8f39f : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/584aa06abb4a4922a5f06b245185961b : SUCCESS in 9m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cb785d8f0abe4fbab658abe60e41b7be : SUCCESS in 35m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c4d010eefaa4751a3d1b804eff78302 : SUCCESS in 34m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/068fbb9404c240b592c11edbbf141ca6 : SUCCESS in 17m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7da05ddaa74a4e49a0a93b2579a8659b : SUCCESS in 25m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/007eaf47a1e24254b37f3f32b8363055 : SUCCESS in 25m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9848f32e08324eb38a581c263cc3d77f : SUCCESS in 6m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1f9b34024a7c432ab8b5d3de7116484b : SUCCESS in 8m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/502a64fb3e7b4a9ea5b220e82eff692d : SUCCESS in 8m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d286b1956e184489b7f19462e6c13aa1 : SUCCESS in 12m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e950bec43bc949549c77fb50e49ef5a6 : SUCCESS in 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0b24ee7dbf6486da58b764a9a14555d : SUCCESS in 53m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5027a9a2b964de4847fb9da3191c5ad : SUCCESS in 59m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c40f1d609d7e46be92e46500f26b62c3 : SUCCESS in 1h 15m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/386b94b3346340c6aef3f032ca99ba8f : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"34e04979b1c5ada3a79b1348d884ca7f67f73ff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 22:25:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b2976e5cabc7372d5ddb5b53b95b365feee0a5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 22:27:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba067300346143078f24aa8c7bc5a8ea : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73f8194eadf6e6ac41e22665546ba85776c67842","revisions":{"73f8194eadf6e6ac41e22665546ba85776c67842":{"kind":"REWORK","_number":1,"created":"2020-04-21 15:04:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/721643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/721643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/721643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/721643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/721643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/721643/1"}}},"commit":{"parents":[{"commit":"f55297cc5737c5f5b866431ad9e9e42116039582","subject":"update constraint for python-monascaclient to new release 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f55297cc5737c5f5b866431ad9e9e42116039582"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-21 15:04:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-21 15:04:16.000000000","tz":0},"subject":"update constraint for openstack-release-test to new release 3.0.0.0rc1","message":"update constraint for openstack-release-test to new release 3.0.0.0rc1\n\nChange-Id: I2d9a7214bd50579b1803af1d0b4cf06ca89d1c47\nmeta:version: 3.0.0.0rc1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release candidate\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I84a5ebc65331f58ee6530c03bd45d76acb48fd39\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73f8194eadf6e6ac41e22665546ba85776c67842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73f8194eadf6e6ac41e22665546ba85776c67842"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
