)]}'
{"id":"openstack%2Frequirements~721818","triplet_id":"openstack%2Frequirements~master~Icbb72e6034d3fe4b3b98ff318cad38837d163571","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icbb72e6034d3fe4b3b98ff318cad38837d163571","subject":"update constraint for XStatic-Graphlib to new release 2.1.7.0","status":"MERGED","created":"2020-04-22 11:05:24.000000000","updated":"2020-04-22 18:34:58.000000000","submitted":"2020-04-22 18:32:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721818-1587580357767-dfe5b5ea","meta_rev_id":"36ae4bcd2d70cc078d8e0bf2628745555b0be515","_number":721818,"virtual_id_number":721818,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 18:32:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-22 11:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-22 15:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-22 16:53:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 11:10:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-22 16:53:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-22 18:32:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"007e04161bae8febad064b1e7c8ac41d2a907193","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-22 11:05:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0af74f2e887178908a4ded885b606777acba3c05","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-22 11:10:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c099bd39aca131be6e27d388d680c3ba62ccb88e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:32:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bbbb01cbd6a492ba730db964baedfb9 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85bb4f5fddbb4abe9c6688a6967d0410 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d2a1152d01846cba851ac2d41fad3e7 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e158ebdd81174e0784d00d5db4be97a6 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/668adc78c82a40888b7931c559c4e2e8 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3c7c069c3064134b5a1e44226d831da : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4de9140628984982b5efeb342e8f8782 : SUCCESS in 5m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/27fe7088b9cf4650ae8ed0bbf9e9dc8a : SUCCESS in 11m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/961a62f36051424dbcd4272265bd32ec : SUCCESS in 12m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a034a9147926496d8cfe3cf3720dabf8 : SUCCESS in 6m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e34556fb34284baaa5ffe90ad99ef207 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d41bac6532247f2baa772d1af9ea180 : SUCCESS in 7m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c656a532fdce42cabb2d0e470c0d3a70 : SUCCESS in 15m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5aeeef1d0ff84c21adc91b3fc826dfde : SUCCESS in 30m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/92b0ac8958b144f28392eb61ff41aad7 : SUCCESS in 12m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cbd13e74305840f98db88a0887ea9dab : SUCCESS in 13m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d782a84bf2f341759cd49bc34d591c57 : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2380b93f26e643afbe12bae31bd7c9a0 : SUCCESS in 5m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/05995d5038ed459bb6e4a8b6162f3105 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e676192eb594907aa22d820424e2b6e : SUCCESS in 8m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b79533de39c24c0e831aeae075a35e5c : SUCCESS in 12m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7010b4466de945629f9ae1cebc95c310 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a46348f756234a13b8ba790f66a18d9c : SUCCESS in 54m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9f3d40107034b33880597968336a8d9 : SUCCESS in 1h 15m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12233c5edb0f42be8429a09113fe19b2 : SUCCESS in 1h 21m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/87158203510e4d8594f0e3aa5366dc23 : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2538202af41d72a09399b23e88e18e5d92c26c5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-22 15:24:31.000000000","message":"Patch Set 1: Code-Review+2\n\nanother late release?","accounts_in_message":[],"_revision_number":1},{"id":"38712c7153327019e8b0465a011231fdbabd62b5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-22 15:26:52.000000000","message":"Patch Set 1:\n\nYeah, this was an independent release, and my fault I didn\u0027t think through that it is used as a dependency. Maybe we should check with the team to see if they need this for Ussuri, or if we should hold this until Victoria opens up.","accounts_in_message":[],"_revision_number":1},{"id":"51343a76445b1e1db9cd1c4d2c5d5efd7e801bfc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-22 16:53:46.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bc46c17759f356b6771951fe6b1857c8263ae71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 16:54:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d1cd753f5de730b8af9f76c55f115e713756c97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 18:32:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c372693b34813d72ef2a6afbec279194540b25a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 18:32:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9316d498c719461aac99dc1739b093d3 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42bfe8b33a2c4b54b5de9b9bb6351f49 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/592bd9fbff8a4a1c91b5087d60204ed5 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58906f54f7024d659d1d9cbf8bcee05f : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b0c62c674c248768eda839e95f390ff : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38711b8cb00a40dea2a79b767a859731 : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/78fc2e0068fb42c7abb483fe1b34e3c3 : SUCCESS in 11m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/61ec1542f9d14c8dbb239d0b611d5beb : SUCCESS in 12m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/663ec9947d45429b800e174164c568f3 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/891a10e8c90548688ed799e42b098466 : SUCCESS in 12m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc6418198f2249d9ac6c6976df9ffc53 : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c001fc12ca6e462d945b051b7f712225 : SUCCESS in 16m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec92219d52e34a508962f98a146e31f2 : SUCCESS in 27m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eed856379db041f2b718e64580c06f91 : SUCCESS in 15m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/29bb772153d940218a889816386835eb : SUCCESS in 16m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/792df4ee20b04ab69223025f9294fa8e : SUCCESS in 25m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bd79fc167d3640f7acd3dcc17a318bf9 : SUCCESS in 5m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cfa82e81b2f64af092964fee36a61aef : SUCCESS in 7m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a8d165dc6fd4ae6a755da2f944954ec : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/61727350642846f2bf7eace4acbfdc05 : SUCCESS in 12m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6533f08de45542d9977e33ee1c3edc62 : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05da193d7b594545a412e27ace323394 : SUCCESS in 56m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ed6122abe054f5b95bce8714934ce5d : SUCCESS in 1h 23m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95836ad4d6e24313a4dbc803ffc65b22 : SUCCESS in 1h 21m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0d6bbc92e1df4654b92f83410ef7361f : SUCCESS in 1h 21m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5d5c600c1b08d0bf5999b9755c506237a8f047f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 18:34:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9adce558138143ff8405ba83fb215ff9 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20e4c3b3a50d242418c0dd741d4eb1b517cdd3b5","revisions":{"20e4c3b3a50d242418c0dd741d4eb1b517cdd3b5":{"kind":"REWORK","_number":1,"created":"2020-04-22 11:05:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/18/721818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/721818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/721818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/721818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/721818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/721818/1"}}},"commit":{"parents":[{"commit":"f052a629a5c626a929dbec50d99ad7e393e5bdff","subject":"Block confluent-kafka\u003d\u003d1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f052a629a5c626a929dbec50d99ad7e393e5bdff"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-22 11:05:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-22 11:05:22.000000000","tz":0},"subject":"update constraint for XStatic-Graphlib to new release 2.1.7.0","message":"update constraint for XStatic-Graphlib to new release 2.1.7.0\n\nChange-Id: Icbb72e6034d3fe4b3b98ff318cad38837d163571\nmeta:version: 2.1.7.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Commit: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Change-Id: Id8b107be0fcc57c5049ffd9507a19d865232a4fb\nmeta:release:Code-Review+1: Eyal \u003ceyal.bar-ilan@nokia.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20e4c3b3a50d242418c0dd741d4eb1b517cdd3b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/20e4c3b3a50d242418c0dd741d4eb1b517cdd3b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
