)]}'
{"id":"openstack%2Frequirements~721822","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Id52265991622a76d5b8cb1794208a3f91c5a6fc3","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Id52265991622a76d5b8cb1794208a3f91c5a6fc3","subject":"update constraint for openstacksdk to new release 0.36.3","status":"MERGED","created":"2020-04-22 11:28:48.000000000","updated":"2020-04-22 23:14:58.000000000","submitted":"2020-04-22 23:14:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721822-1587597298790-653b66dd","meta_rev_id":"85996de7ce712f430a5b9a2969d87b1265051126","_number":721822,"virtual_id_number":721822,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 23:14:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-22 15:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-04-22 20:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-04-22 20:34:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 15:24:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-22 20:34:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-22 23:14:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f57f3f3cd3199b6da967e2c1b7614760ef6a5eb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-22 11:28:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38c1df60d8e633b8ce54de31caeb2985247d8331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 13:23:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/29546b711ecf40b8a9df781ed16cf1b7 : FAILURE in 6m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89bd6f4587ed482883ecaa4b752481da : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/138d3eaba7d6483ca33674d4be944fdb : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e74d1dd8dcf48d28a3a69c54cfe7ea4 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91451e433b4c421e977e975f1525619e : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fab7333e9cd4718b8fab4ee34067732 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/244045c2afad4a949f395c52c6996cf0 : SUCCESS in 4m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/af69fe3708fc45ef87afc49170cfe50e : SUCCESS in 10m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a26562374804dca8763d09ca64b6286 : SUCCESS in 9m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e5c2d24df3284db0827e51ccc3ce796c : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5fe250ae460240dabe83094fe050b08d : SUCCESS in 10m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cf936bfc8e95445c99d8ff8928120ffe : SUCCESS in 5m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8670bf94f75f4e9f9d729470aa72547b : SUCCESS in 6m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6fce7f8d18b34dd29a960acf55af61bf : SUCCESS in 10m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2282d92dd097478fb3ef9573b49b6ad0 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/239611972c164a80b5ffc9f18b3855d1 : SUCCESS in 8m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/dc14852a74f846ac99f7e02b64e4d643 : SUCCESS in 14m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0fdf1aaa1ba94e6d99b53c7082dfa4a8 : SUCCESS in 13m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/de7d4549694641128c41dd9cb356bfbf : SUCCESS in 23m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7270ab767af4ff6a4017582dd9dfd1f : SUCCESS in 26m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4220bfeb5c674878acd56426148f8752 : SUCCESS in 14m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/249ab87f2a694c438a8a80eb2bb69734 : SUCCESS in 16m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7931e210d16543b2bb73826835ee1397 : SUCCESS in 17m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e44c8cd872be482d8eb3a417f961bcc1 : SUCCESS in 22m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4db606acb524c46ac987b814a57bc0a : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cf296869a92f4af8a699839a04428ba8 : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cfbd736c3bee47d6ab9e477af1e14f9b : SUCCESS in 10m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a7e25d8674c40289e9ae14267c583e1 : SUCCESS in 1h 20m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ddce41acd4c34740b00f4f40c2352c1d : SUCCESS in 1h 48m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6a0c1e5a7d4c4a238f3511320d77535f : SUCCESS in 1h 15m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d97a50c7944f40656bdf4e5c109844e7add95953","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-22 15:24:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04117e0229801cbebfffc4c609eaaa47ed622b17","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-22 20:34:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9aeba665fbc50d2fdd251e8f45acafefc9d910e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 20:34:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fefead303ccf230b8bfcf3f77827bd7da21ead4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:14:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"55b6ebd10f950e6a067b844052b3da1dfa8a425a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:14:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7957a69873f1430f84014c490af258b0 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e52625a4e3c1438eb6c862a0d2caecc9 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6294dda29cc0496e85633bacdcf26c6f : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0879e7ac0dc04a6c912ed7fac3d557d2 : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea9937f746ac467b8adb06edfce95d49 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d875670b7a754f7683cd81d538f6323c : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0d265a2557874e40aaf55507f4703cbf : SUCCESS in 9m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/276049e773f143b2a34455b528e97936 : SUCCESS in 9m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8c9c191ef50944bfbeb2fcc7848bf4e3 : SUCCESS in 11m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aabb862095fa4536888d486b210d3c6d : SUCCESS in 10m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/51ced3c5031f48c98a80a0c426f7d880 : SUCCESS in 6m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fa8187862eb240aaa10d74ecace6445a : SUCCESS in 6m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b3e87ba185ce4794a892c3bcdaac8764 : SUCCESS in 15m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/38d0709264aa4788a56c59e65d7cd192 : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d75fa1a9ebad4de796737a346b0b9044 : SUCCESS in 9m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/066c58f0206d4df0843a25ad1203413f : SUCCESS in 13m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/adc909a125da44999476ef039687acb6 : SUCCESS in 26m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f05cb77cf31f4286aa266cdc4a6c862b : SUCCESS in 22m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/49ea835404504c1980f83b1ba190f21c : SUCCESS in 28m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bfea8e52a3fe4360bace8af8c79df2be : SUCCESS in 23m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/88d97b5f40fd45c6bed80c4ae81253fd : SUCCESS in 15m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/329adc8d779243eeab5b2fa124feb7f7 : SUCCESS in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e3f759348ff418bb69b5db46c248e8e : SUCCESS in 23m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc41f220b5a14c7caede7fb455edbc59 : SUCCESS in 6m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e267438fa6034d6a84a3e110b62e81be : SUCCESS in 26m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9633d5b31d434fc087a986ec6a259673 : SUCCESS in 11m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc0e182a612c47b295ec714f28e9ce76 : SUCCESS in 1h 20m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c960d929b960463ba09636eee91895f4 : SUCCESS in 1h 33m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/72851fb5d0da48d599443c79d17031e7 : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b7795f1c11c6294a75cfa0e53d511a2242d1163","revisions":{"5b7795f1c11c6294a75cfa0e53d511a2242d1163":{"kind":"REWORK","_number":1,"created":"2020-04-22 11:28:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/721822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/721822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/721822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/721822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/721822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/721822/1"}}},"commit":{"parents":[{"commit":"95257c78e7ba2b4c571b0449954e46c72f63d201","subject":"Merge \"Update upper-constraints for pyScss to 1.3.7\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95257c78e7ba2b4c571b0449954e46c72f63d201"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-22 11:28:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-22 11:28:46.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.36.3","message":"update constraint for openstacksdk to new release 0.36.3\n\nChange-Id: Id52265991622a76d5b8cb1794208a3f91c5a6fc3\nmeta:version: 0.36.3\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: Ib34e5c188d5385ef4d1afdf5612310544df27f57\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b7795f1c11c6294a75cfa0e53d511a2242d1163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b7795f1c11c6294a75cfa0e53d511a2242d1163"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
