)]}'
{"id":"openstack%2Frequirements~721956","triplet_id":"openstack%2Frequirements~master~I50e0c16f796ab88b6301cc2c9344daa9ccecf287","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I50e0c16f796ab88b6301cc2c9344daa9ccecf287","subject":"update constraint for ceilometer to new release 14.0.0.0rc1","status":"MERGED","created":"2020-04-22 13:35:04.000000000","updated":"2020-04-22 23:16:26.000000000","submitted":"2020-04-22 23:15:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721956-1587597300042-354c625b","meta_rev_id":"d810eb29883b0e39250ea25f2a7816225e47cfff","_number":721956,"virtual_id_number":721956,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 23:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-22 18:31:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-04-22 20:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-04-22 20:47:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 18:31:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-22 20:47:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-22 23:14:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6bd5ff0b184ac68d1fdf991bf11d8031b3f87d5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-22 13:35:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c9eb2cc8f0c77396247aa08f1a4305f27c37ae4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-22 15:25:08.000000000","message":"Patch Set 1:\n\nused as lib, late release?","accounts_in_message":[],"_revision_number":1},{"id":"d2eaf20886c9c30e97c13c8965ec59d65f43f26e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 18:20:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482c08b59e3c417dae5ddb9a9f9f3dca : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e238a54bb56c40bbb3dc3fef22ebfceb : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf236e83f2554ae09198e235ad607d1f : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cd264f3d09d4db0afac3ed13ca7b477 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e23a90ffbf34545bc839bb73654c80f : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15b48380808d49318e7b1fa93873e030 : SUCCESS in 6m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6258281039a54d83a344658358ffafa8 : SUCCESS in 7m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70e45f7bf29f489aa09089d2acb3831f : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f1b5bada1fef4bab9f9cccf528ab0ef4 : SUCCESS in 12m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c13d2796723945bc86dd54b2e62abaae : SUCCESS in 7m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cf100ed7c3c54187b2a4b96c269cc86e : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38df15414b5a44a5a8cbe00352b8110c : SUCCESS in 9m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f063b18b08084f4ca6dff321049c4b74 : SUCCESS in 26m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cbf6760572d94088b2e7cf02d9a029d5 : SUCCESS in 29m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6c3891065a2c408497d05925918ee5cf : SUCCESS in 16m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f3ad72401168438fbbb40fab86291294 : SUCCESS in 15m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e0439045df549189599f8f8c285e26d : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/156b1cfcaddf4e61ac11982d9dbec26e : SUCCESS in 8m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f7ac797bf5f84d288d473bad53b68064 : SUCCESS in 8m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c1bdb6d15344b2f89724c3d5fb4c161 : SUCCESS in 9m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f5af59445fcb4cf18a2a648c79cca590 : SUCCESS in 12m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1d95e92160ee4d2cbe98ac095d4169c7 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ed39362482a437398080c742a0a3493 : SUCCESS in 55m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/45a732b63a30402484e99b312c068180 : SUCCESS in 1h 22m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05b112011ddd4ddfa453fa928ff945d6 : SUCCESS in 1h 40m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0371fcc01ed7455db30014beb8ad723b : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"569b55b9d1c7091a59ca1e6fbdc09a48fa75369a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-22 18:31:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0dc60c927754beba52523fef811af076d7ce2a5f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-22 20:47:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d67040f44e17a04f33ae580121c7217c8432a33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 20:48:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b04dfce6ed32d47075d4c4c4e2ac51df41d300e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:14:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5746945f9cca4fe7b6b7da9a1ceb7601 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0c32f7057245409337046dfece1eee : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f87451af02b4e9ab2e2c0951e547aae : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c20d636141a64ea2abba478475595908 : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23e01bbd83104dabb0529407da9740ab : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac0a6b0e83b94cc5a5319afcfb4921ec : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/38df0b1fbc3644aa8752f66620dd4356 : SUCCESS in 9m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2ae88af0e8ca465da998aa9d5a11949e : SUCCESS in 10m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b9e7f96b0c5c4c9583c725d5d50f3860 : SUCCESS in 6m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54da0492030f48678993dd30e337d7df : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5da83d7809842689cf54f7139eda4d1 : SUCCESS in 7m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a462aa3fab746f9943f59fa4c54e698 : SUCCESS in 27m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9359c8abf2a04649ba16e5cc1b2a7a1d : SUCCESS in 31m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/52f7b9e4eb4341cda02e5ed4d58fd780 : SUCCESS in 21m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/59f89955b2054bb3a8d71e181e9b831c : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35e0099196a84426b9d14531cfdee912 : SUCCESS in 18m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ac34d33ff6094f2fa1512f39a1883783 : SUCCESS in 7m 39s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b1519d32986b4281ade131937a44b66b : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/061f7216f6bd45049b7908a27ef5e759 : SUCCESS in 8m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fd81cf1c9a714348848c3614babbf2e4 : SUCCESS in 12m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f57967745c0c4839a85a96e67e77ca10 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fc793711ac64b2a9f3f8785fed0b855 : SUCCESS in 55m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de484395e9e849fb8de1d8345ebb0402 : SUCCESS in 1h 21m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae9e0aa65ea5475c9851d437f38affde : SUCCESS in 1h 26m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c633ceb6a72d48988a433909ca94ca1d : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3b8a292cc23536ba52c2ab67d52fd3dacf5b4827","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:15:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"027e0b1c1016426df820a061c76245a251d38024","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:16:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb8c27cc49524b68ba0958413ebea35c : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d01d684aae04075b97b842bffe33977813826b8b","revisions":{"d01d684aae04075b97b842bffe33977813826b8b":{"kind":"REWORK","_number":1,"created":"2020-04-22 13:35:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/721956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/721956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/721956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/721956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/721956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/721956/1"}}},"commit":{"parents":[{"commit":"f052a629a5c626a929dbec50d99ad7e393e5bdff","subject":"Block confluent-kafka\u003d\u003d1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f052a629a5c626a929dbec50d99ad7e393e5bdff"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-22 13:35:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-22 13:35:01.000000000","tz":0},"subject":"update constraint for ceilometer to new release 14.0.0.0rc1","message":"update constraint for ceilometer to new release 14.0.0.0rc1\n\nChange-Id: I50e0c16f796ab88b6301cc2c9344daa9ccecf287\nmeta:version: 14.0.0.0rc1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release candidate\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I24fd52ff6f16b1cfbfc0fac2fbfa4313d80a141a\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d01d684aae04075b97b842bffe33977813826b8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d01d684aae04075b97b842bffe33977813826b8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
