)]}'
{"id":"openstack%2Frequirements~722216","triplet_id":"openstack%2Frequirements~master~I29d8561479ebd2ed84c497e11761b75ddabf3fae","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I29d8561479ebd2ed84c497e11761b75ddabf3fae","subject":"update constraint for tempest to new release 24.0.0","status":"MERGED","created":"2020-04-23 07:22:50.000000000","updated":"2020-04-23 19:10:23.000000000","submitted":"2020-04-23 19:08:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722216-1587668890216-602cea05","meta_rev_id":"57c747c5f51c46c646a9b646ccb2959329ec6f6a","_number":722216,"virtual_id_number":722216,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 19:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-23 12:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-23 15:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-23 15:45:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 12:18:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-23 15:45:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-23 19:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41210f6b07d51b1034f69e2123d72838db659cbc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-23 07:22:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"924fe2adcce1dbe5cc9ff0209391d7f23cede45b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:26:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a643c78fd8d4ba192c7a021037f18a3 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7433308c174ef785ba1f57d0e41b65 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95636b5ed5de4f29a09c94085304cdc0 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eebe3d230a1743a88c0918d484519376 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/439cb18010f7482cb25e5656e118cf5a : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfb5dd75571048729c8e5c8e13b62112 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a267ed69852e4e5e8c0169be774b786d : SUCCESS in 6m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/31b136a2c6cf4ae4b1615314dd981759 : SUCCESS in 10m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3fa679de30014aba9a08224bf930d35e : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/085d728089524c2b9081409d19e35cd4 : SUCCESS in 9m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9d27f608d3584ccfb0e210fd22fa177f : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70e586fa38ad465c8f0fd5ea51297f39 : SUCCESS in 10m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/62b8d1ee0d914aa4b35f29dbaca8bca2 : SUCCESS in 18m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b62e3cefd92d4e248c8b8036baee6d99 : SUCCESS in 27m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bb021e04989d44e69647ec291bd27e23 : SUCCESS in 13m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3288ee25fbaa41aea9cd127c2f335945 : SUCCESS in 17m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d4328085d32a460ca2ca55622c488050 : SUCCESS in 19m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/767eb3b6d571407a97f74749c48ed295 : SUCCESS in 5m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e3b235fd5a894091a879d2f579d34604 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3942c04bffe447a088d7aa10ae4379f8 : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b5071f64b57d45559b32781c98f28431 : SUCCESS in 22m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f397a5f1e81e4bacad194afb3300c2bf : SUCCESS in 17m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f633ae4bd2540ee852341c0b003e3d1 : SUCCESS in 51m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b13d1a4bf8b44c8bc16aece4cbaa34f : SUCCESS in 1h 22m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9920b62053e64bbd99f9296ba0932609 : SUCCESS in 1h 28m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f5fc7220e0a4cd98c04e006ee3d4db3 : SUCCESS in 1h 18m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3c0cc6713752026d892ffe9c1ceb769def5cef98","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-23 12:18:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"88023e8b9149acbc358576e404a0834f535a640c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-23 15:45:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2c1d7af1de9fac993ead2afa949ac87c388978a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:49:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebc09ad80e4ac6ce54cdf14caa577587c1141c5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 19:08:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc6d3c73c05418e91e638726465cea4 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e7f6db2a93f4489b9fd385f9265de3c : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85c039a8e9ed4271867ff209348e41ac : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0cc289f266e402c9284138c7eb36722 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de168a618c8343cbac78119530b98582 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79cf5f6e908e4219af3e5c1190082daa : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/66cd59bd7d5641a4ba902ffb11289112 : SUCCESS in 9m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9c9af75cb86c4dcf98d8cd6b42b1562f : SUCCESS in 9m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1163e27dea4e4605b4c2151781d98cf2 : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7b0d37d74f714a26b18360f9e7b5f824 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1130206d64034b2ba257c3f0c74101d6 : SUCCESS in 9m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ac6aa3aba528427cb68cf8d9d755cf9c : SUCCESS in 25m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1b1547f76d244003bc03d6dd950d6506 : SUCCESS in 24m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f72672a8d34243a3a22aca0a99444df3 : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b6762540d96c495884410eb60453f75f : SUCCESS in 12m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97515d7b06bb4933ab6f6c23955f1da5 : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/136a4220c38a426fb5ffdcb79e6d255c : SUCCESS in 5m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/65970ef6ddd04cd890272cc89515b7dd : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e26c6c9deb2d46a1b8d9b71161d92193 : SUCCESS in 7m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0bffff50735442a2a8cbfe8dd1275e64 : SUCCESS in 10m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/45b7bafe1fcf41628a8da64cd7ffb8ab : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a599e8c6a9314839a3a2d08215429a91 : SUCCESS in 55m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/21d6e48d714d4b8ba4adf7312bc618aa : SUCCESS in 1h 16m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c801f301acf430d83a06c96636537f8 : SUCCESS in 1h 16m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a25eee4f04594a9abcaac04b223de3a5 : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9940f14da0ee94933ae7b3a5455fbf0c91f5725b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 19:08:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e57b5b000776538c5b88599219fd9b135b51dcb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 19:10:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace5eb306ecc4bce86a9aae3feb8aa7e : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc74822f08abd3598ff95f7703576a583eff7780","revisions":{"bc74822f08abd3598ff95f7703576a583eff7780":{"kind":"REWORK","_number":1,"created":"2020-04-23 07:22:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/722216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/722216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/722216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/722216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/722216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/722216/1"}}},"commit":{"parents":[{"commit":"cd6b6f36247109a4d80853ec4cd42de8a8279971","subject":"Merge \"update constraint for ceilometer to new release 14.0.0.0rc1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd6b6f36247109a4d80853ec4cd42de8a8279971"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-23 07:22:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-23 07:22:48.000000000","tz":0},"subject":"update constraint for tempest to new release 24.0.0","message":"update constraint for tempest to new release 24.0.0\n\nChange-Id: I29d8561479ebd2ed84c497e11761b75ddabf3fae\nmeta:version: 24.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta:release:Change-Id: Idacc0ca037123685438bf55e4c91fa9276d33135\nmeta:release:Code-Review+1: Masayuki Igawa \u003cmasayuki@igawa.io\u003e\nmeta:release:Code-Review+1: Martin Kopec \u003cmkopec@redhat.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc74822f08abd3598ff95f7703576a583eff7780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc74822f08abd3598ff95f7703576a583eff7780"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
