)]}'
{"id":"openstack%2Frequirements~724316","triplet_id":"openstack%2Frequirements~master~I6effb2ce2813ad5786ba964ecadf75b4f5cc6bb3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6effb2ce2813ad5786ba964ecadf75b4f5cc6bb3","subject":"update constraint for virtualbmc to new release 2.1.0","status":"MERGED","created":"2020-04-29 12:56:34.000000000","updated":"2020-04-29 21:57:45.000000000","submitted":"2020-04-29 21:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724316-1588197273620-5004bf0a","meta_rev_id":"1821391bb79a439ba6cc262341b230235c8b6888","_number":724316,"virtual_id_number":724316,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-04-29 20:07:28.000000000","_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 21:54:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-29 20:02:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-29 20:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-29 20:07:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 20:02:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-29 20:07:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-29 20:07:28.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-29 21:54:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f42863bb676ed2dce90915242207a3a6e69aab1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-29 12:56:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43c0a0f359988596c7f7f71ad79787ddff385e05","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-29 12:56:59.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"ubuntu-bionic\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"4b20ce4386ac1f42ca80f50ee87f0c0c7de11c34","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-29 13:03:04.000000000","message":"Patch Set 1:\n\nHmm, I don\u0027t actually see this in any project\u0027s requirements.txt files.","accounts_in_message":[],"_revision_number":1},{"id":"c1d152cd556af4c3363d8f2c4ce4915a08bfe280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:10:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b399ba850caa45e9b2308bf81591b601 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee0a59a63234131b7231e076251f0ea : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bde4794989f4638a6aca8b38937697b : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2db2e444994b4165a30cee820a3041c7 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2d2a3f6e1514dd389a59b3985af039a : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e69686d583bb4c7c86834bf01077f77e : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cce10317806845049cc086047dd28518 : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/136c47c670334195a13a478090955231 : SUCCESS in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1ef819d6d58e4a00b55b0536b96a271b : SUCCESS in 10m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8cbbb9107dde475c8ead4f1d5ac9433e : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bc876e82bf46434782889998237f89b4 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca035372a7bd445cb6e1f2924e34afd5 : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d4716ce144bd4f929328b582084799f5 : SUCCESS in 14m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0ca6bf8150474130b83432a251e0e2db : SUCCESS in 24m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0fa60371fc7f4dfd8ffd97d7c31c0de1 : SUCCESS in 12m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a4f33630fee34331befd8cf35a023f9a : SUCCESS in 13m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86dc0b9f86784ea4b1e289cc2c5cebda : SUCCESS in 15m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/49c6606f78ff430aa9731356f9693039 : SUCCESS in 5m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/14e4576b80014ba2a282750c17f15d6b : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ce181018eef45e68cc17352d09dce0c : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9bccfadeaca54068902cbe4ebb404a28 : SUCCESS in 9m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/31909365c5a9417fa6baffb2011e1de5 : SUCCESS in 9m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11f14d1a4b724f32b9272281df700da8 : SUCCESS in 54m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bef63663b61645a6bb453a842916e336 : EXCEPTION\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e81fb61a9c54b7e88ac88c93a8edd1a : EXCEPTION\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/673fcd845e40459caf62a8e7e6e5c76b : EXCEPTION","accounts_in_message":[],"_revision_number":1},{"id":"795b2e8276bdd44b24d67a0fc4aa40696efcc7a9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-29 15:35:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e811257e1403d802ea25dde9c7e17d972ca116d6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-29 15:35:58.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"ubuntu-bionic\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"dba66742468566050a000aef86c454e88482151c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-29 15:36:35.000000000","message":"Patch Set 1:\n\nsurprised it\u0027s not in ironic\n\n| x/flame                | upper-constraints.txt   |  128 | virtualbmc\u003d\u003d\u003d1.4.0                                      |","accounts_in_message":[],"_revision_number":1},{"id":"2ca569c542ad5d5cc4a6bcf90dec40f6599f5d1d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-29 15:37:55.000000000","message":"Patch Set 1:\n\n\u003e surprised it\u0027s not in ironic\n \u003e \n \u003e | x/flame                | upper-constraints.txt   |  128 |\n \u003e virtualbmc\u003d\u003d\u003d1.4.0                                      |\n\nThat explains it. I was only looking at openstack/ projects.","accounts_in_message":[],"_revision_number":1},{"id":"5849d1ce25c93193e030d628faa43f3ac2025f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:14:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff085608a4f4cb49bc2146aa53815bb : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d449c9fa29ac444389b0fcb9b12de1b7 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4e7421533164d2eae7900c2e0a43844 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61df13c46ae548b2b960ba3e8e98f12d : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/012112c3cc47408e9b51ee6ab2826490 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f090f87b71c0444abc44734a69bb9ac5 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36eaeb8e2ffc43ecafe4901c6de74ee4 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71d9b23226d44a24a38cbe411bcefd5b : SUCCESS in 11m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/71dde03f9ced4c8abdb2a86e7d63e12b : SUCCESS in 12m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c95d3b0edc3d4cb8b2e744029095f02e : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fbaa2524bf1e4f9ca006174f6d97bf46 : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e5598889f2748a2a948721265cbd945 : SUCCESS in 8m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f6d334d9dc3247718b35c7b49ea04c62 : SUCCESS in 19m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5ed3bfdb7f0c49e2b1ee07d5eb424d54 : SUCCESS in 23m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dcd099ecf8d74a1881e9e9dcdf1564cc : SUCCESS in 11m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c6a9b4c5cf9c465aa91732889350df0e : SUCCESS in 15m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/edccbb2750d24623a4d60ff8a1e22f83 : SUCCESS in 18m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fcfb85df6fd8404d9606d757d99cfa30 : SUCCESS in 5m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9055fe2eafa740e4b89776d87e0c592c : SUCCESS in 7m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c59814cd9d84dcabeed9ca2ce43e6df : SUCCESS in 6m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2451b8dfbb384821a4c8db5e1415b812 : SUCCESS in 12m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/940cff50c87d44a3bb4d2c25cbe6a494 : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f1d38c646b248bca04022167a7d53be : SUCCESS in 57m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e156d28613d4401b334f92ae8b919f4 : SUCCESS in 1h 05m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09cee757b2c04c5c969ac3bc32ceabc5 : FAILURE in 1h 09m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ddccfd9893944808ac44a5133ebdb26b : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4db9b61160f6a695498dbb21c49e29b615ab74bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-29 18:03:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b02cec7b7f61e6fe9a9fc78df632ca28656507ca","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-29 18:03:50.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"ubuntu-bionic\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"4040a6f972b94977948c2909a9d4a6c5659e97cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a4a34e57db4b1693caa82748599195 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77fbfc5b504d4c1e9027620e16020f52 : SUCCESS in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/020b868963514d81b04a6844acc6efd3 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/727a20d0c9d4452b81d0543e31967b0a : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eaaa414454674deb949ca79b8194d7ca : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d417b8d2662f41238f59315b49a151cd : SUCCESS in 9m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b650ede7e20c4492b779a2f90ba8e111 : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6999bdcbac7d4e4589c531333b9ea301 : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ee5286d568834b5ea61d235273eaae54 : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92a094c218a8415aac334b6c1835f2b5 : SUCCESS in 13m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27a0ff88c52140c9bb9d853bc9e8b800 : SUCCESS in 7m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/47ed7160a4194fe8b2094ecf9cb998d7 : SUCCESS in 14m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/803a19387fac4c3ba668c46c36a67195 : SUCCESS in 29m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dea0b463c53a461c90473d2653b1f407 : SUCCESS in 14m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/43ea880d9b4b414bab550a95f987c453 : SUCCESS in 15m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff9103e9acd14fccad5d9e4189217155 : SUCCESS in 19m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/38745bfd1f294524a18744ce2b0bc343 : SUCCESS in 5m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/46b7f5b5af814516b994cb4f0cdf2806 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4dd807b04d604efbb3098e8d88a6191e : SUCCESS in 8m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2ed9771e7ffd495da315ca72d46a35c3 : SUCCESS in 17m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8c320e80c7cc4f82a18f285aec3330e7 : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4af256bdc2784302ba5a6e692f261366 : SUCCESS in 54m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/797e8d7938a74f519e3130d14af60d7f : SUCCESS in 1h 00m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e61eb631c1194f4ca9fc05f5c6e01762 : SUCCESS in 1h 40m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/91e3420467964015afbd564f8f45ce9a : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1912fee902795445c325a4474ef0dbf936c2383e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-29 20:02:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d172ba2e63fcf89dee6a18305233538d0d8ef0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-29 20:07:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e943785c7f59a155d14ebf5d2453f06fddb06743","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 20:07:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76e92aebf016d7c0508375f33f783f6d6337bac1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-29 20:07:28.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"ubuntu-bionic\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"f302167e5ee901157da86f4f38c8063cd5c670cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 21:54:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e22dc9787d8490bba1ce35ed7e1accb223187a03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 21:54:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971b51253fdb49d3a06e9e0a6cd5a703 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8538faaa423452e8dd2417bd0dddd4e : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff1a2ada479a4feb851f4657106cc910 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/507c3fdb19514975a8c725a5003b7dda : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a2e0503c4ad4c65abb837130247edad : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66fa0f2694a2404da7c6dcfa51a99bdb : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd351be79ab54d3ea735fbc739f1b3e9 : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/677d6543487b44eda09568168c5d063f : SUCCESS in 9m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0e7af16688bd4a23aa70a811d58456e1 : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a403b9ac29e440b7bd9f761c17a4bd7b : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36e1b3dba29a467c90d565d49b77da0b : SUCCESS in 10m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0137bcb1c07649b08523b73ef1f8e8c4 : SUCCESS in 12m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e2b6d693fde840f0ae8f0232c7b60b6d : SUCCESS in 27m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a428210e1cd48858f38ae57c31f9fe9 : SUCCESS in 15m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bac3e30327aa4ee398d9cb9fc31417ad : SUCCESS in 12m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14cebed61e21495084cc60b2ea4af2ca : SUCCESS in 23m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b33ee9b5d0b04d5e8591dd5c85343092 : SUCCESS in 4m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fdd91a4d921144bba25f7badeeea8bcd : SUCCESS in 7m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0856ea8a6acc40599cd1319986a3adaa : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/78cb31a687c242058d83d2ef478511ec : SUCCESS in 19m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4b47817972f24fefb4289769627fa121 : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80790256e1a64576a3ccf90b2c147af2 : SUCCESS in 54m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed02b3ddc71e4ff8a180137d33e54165 : SUCCESS in 1h 01m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76fc0ae6788441af9b8bafeeb5ac12ce : SUCCESS in 1h 15m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e93884a4a6d4c38a916f8f8d2586880 : SUCCESS in 1h 09m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e7155e7e8cb1cba5735c3d6140b67605fe5ffeb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 21:57:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a00e43b8e84128b9eab0d1a47e3840 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"012faece5de74852aeed3d8db6cc6647263e76ab","revisions":{"012faece5de74852aeed3d8db6cc6647263e76ab":{"kind":"REWORK","_number":1,"created":"2020-04-29 12:56:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/724316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/724316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/724316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/724316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/724316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/724316/1"}}},"commit":{"parents":[{"commit":"b85e113e6c2388482ffc8b778c8757ba49ad4615","subject":"Merge \"Add validations-libs and validations-common to openstack_requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b85e113e6c2388482ffc8b778c8757ba49ad4615"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-29 12:56:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-29 12:56:32.000000000","tz":0},"subject":"update constraint for virtualbmc to new release 2.1.0","message":"update constraint for virtualbmc to new release 2.1.0\n\nChange-Id: I6effb2ce2813ad5786ba964ecadf75b4f5cc6bb3\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: Id01d7fdc0e8d54ba9c33930463408993a22bed49\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/012faece5de74852aeed3d8db6cc6647263e76ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/012faece5de74852aeed3d8db6cc6647263e76ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
