)]}'
{"id":"openstack%2Frequirements~724551","triplet_id":"openstack%2Frequirements~stable%2Fstein~I1e2d64f02b9bb080e98f8c3c9def081fa10a1e33","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I1e2d64f02b9bb080e98f8c3c9def081fa10a1e33","subject":"update constraint for horizon to new release 15.3.0","status":"MERGED","created":"2020-04-30 07:11:40.000000000","updated":"2020-04-30 22:20:10.000000000","submitted":"2020-04-30 22:20:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724551-1588285210203-39316be2","meta_rev_id":"1feda0eb8aab9fbdf90293fabc47eefacf94bdcf","_number":724551,"virtual_id_number":724551,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-04-30 20:27:29.000000000","_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 22:20:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-04-30 16:01:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-04-30 20:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-04-30 20:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 16:01:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-04-30 20:27:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-30 20:27:29.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-30 22:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60921ff3b3a5909616d2fc0a169e64f051c482ea","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-30 07:11:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74efc4d0147c0f4fc94f1489252ddd652c0456bb","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-30 07:12:06.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"ubuntu-xenial\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"4750c3423cadd85fd5aea7363087cf991ee4b546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:52:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7c775eb4e8fb4ca085e0c40e098baf89 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2805e8561c0c4824b1b36663b11592ca : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42a960f6fe694a38898b5c4f22b88854 : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/62ee3eb3e8174b67a00554daa77f6c9e : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d43c175d51aa465db8c1a46c6fb3d258 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/453f439206044b9ead2953e0221aca14 : SUCCESS in 6m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f77c5023c0149969daf180143b0c1c2 : SUCCESS in 6m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/54060ace959c48b4b9e68b978819d060 : SUCCESS in 11m 55s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d057793a7cbc432a9dcd02ccb53288ae : SUCCESS in 14m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef33a45685354a198bb51e6ae594f753 : SUCCESS in 13m 32s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/326c373cf1aa49d7a3fd933fd347edeb : SUCCESS in 16m 34s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9aed8893b9de4252b14113f3007e662e : SUCCESS in 12m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ebe915c45e4a4353accec19c8efb7d89 : SUCCESS in 7m 55s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/34352f1902e04819a4c83a44b2ae1c63 : SUCCESS in 7m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1c7e9ede72ec470581b842219016ff78 : SUCCESS in 11m 20s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/b14a08b68b4e48eda7ac291c78b76d41 : SUCCESS in 8m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c6cdda176ac463e889fc3fef0abd775 : SUCCESS in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8ef56f193d6e492b920b8c9c31fe977d : SUCCESS in 23m 14s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b9c419c274a24cba81f3f52a038a0f15 : SUCCESS in 26m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/614dd515c89d47e0926762377ac5c768 : SUCCESS in 20m 39s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4ceec20b19904d91afa9f60059520387 : SUCCESS in 25m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8900df107c0f487b929770b122c7e040 : SUCCESS in 15m 38s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f71391dbefa04a31b35def2f7264c78f : SUCCESS in 16m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9d915a41b624a3e96b1dc63bb146cb4 : SUCCESS in 16m 48s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e09e5399ac374eaea6a8892fc102edd1 : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/29124d9130264d90a49f0f3953d30ece : SUCCESS in 7m 56s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/91f83134f899437ba5e5e8c3b8246bc8 : SUCCESS in 12m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6a0faa0a17044bbacf2b3d8e6914661 : SUCCESS in 1h 38m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/55212c2f03014a509b033a6e3a6658d0 : SUCCESS in 1h 22m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/98af04e3058a4f14a5c4beeb534ed71a : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5c1f2a321953147fc9c06ffe246efad7dade4970","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-04-30 16:01:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d254e83c18dfee27816b2b8bcd73c3eedfc84895","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-30 20:27:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b92b032fb1a461fae36d48299a0f789cd051cfe2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-30 20:27:29.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"ubuntu-xenial\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"be64be8a551499e6cebf823c12e84267d38e85be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 20:27:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8572e035d5edf21d73151e755d31dc3e358bdfd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 22:20:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75aaeea88e864309af9eb71eafbe6c2d : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bc8dec92dd14aecafa8b347325aa941 : SUCCESS in 4m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e94106cdc3349b08f3ca0ba3cc96ae3 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/033eaa49c113467d9cb8d4fb84045351 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef269dd857c2426c9c3692f999fc29f5 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74524a7557b84581902f6b48d6562edc : SUCCESS in 3m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/73eef6c7fcdb4ababcadb510ad1d4b80 : SUCCESS in 8m 06s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/0a5e463f92f24621ae95d965ad49d8ba : SUCCESS in 8m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4867f934732f4f5698d2c52ad500d2b6 : SUCCESS in 13m 33s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3cf758ba980f44c1a60a4c1354611aef : SUCCESS in 10m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/02e96a84307044aba699ec5e98c1c249 : SUCCESS in 5m 21s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/37530443c46a40db8a08b1798e697916 : SUCCESS in 5m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6e574bb5bbc848aa84e262d67b2564b2 : SUCCESS in 11m 18s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/791d6791d2b24fa2887bd4b23f2029b5 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbf40bae5f904165abdae778799647dc : SUCCESS in 6m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cbaa472174704bc1beb8ebef4b7c9dda : SUCCESS in 22m 14s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/de32e12594434452875b567acf8a3098 : SUCCESS in 27m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/222d99e4602b4e778f49d94885bfe248 : SUCCESS in 23m 20s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8e09ae5ee2db468ea949b8bed2a7f695 : SUCCESS in 27m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0a9e9a6336a94cc7be878a93eb35e2f2 : SUCCESS in 13m 19s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d20b6da0f52f4ca097e47d46f5843eb2 : SUCCESS in 27m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bfc9f37e7dea42288e5cced7175830a5 : SUCCESS in 18m 32s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/57246f42b9e74e8997c5d12f335fb6c4 : SUCCESS in 7m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c37796a964544c93bfbe6de89746526f : SUCCESS in 9m 33s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e26d1ad50ee04608991e352864285fb9 : SUCCESS in 9m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/88e5b2e1b7084ea494d68df9b5956490 : SUCCESS in 1h 10m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/419586a1b870497a86ca80efffc44864 : SUCCESS in 1h 24m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5d0d154e6cc14425b3877510f31ac1ed : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5fa2a21fb94ed7432996ceeabd5643a1dbe64833","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 22:20:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"881814daf02179eb47be4f3ce248560b2b8199fd","revisions":{"881814daf02179eb47be4f3ce248560b2b8199fd":{"kind":"REWORK","_number":1,"created":"2020-04-30 07:11:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/724551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/724551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/724551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/724551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/724551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/724551/1"}}},"commit":{"parents":[{"commit":"015b04615ed19cc823895d4ef9a6200dcdf21b96","subject":"Update upper-constraints for pyScss to 1.3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/015b04615ed19cc823895d4ef9a6200dcdf21b96"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-30 07:11:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-30 07:11:38.000000000","tz":0},"subject":"update constraint for horizon to new release 15.3.0","message":"update constraint for horizon to new release 15.3.0\n\nChange-Id: I1e2d64f02b9bb080e98f8c3c9def081fa10a1e33\nmeta:version: 15.3.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nmeta:release:Commit: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Co-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Change-Id: I873fa2030319e43edd67d55c3c0c1f3d6aba6617\nmeta:release:Code-Review+1: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/881814daf02179eb47be4f3ce248560b2b8199fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/881814daf02179eb47be4f3ce248560b2b8199fd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
