)]}'
{"id":"openstack%2Frequirements~725067","triplet_id":"openstack%2Frequirements~master~I158a8c449d6da6d9e45eba645cf0fd04320a972e","project":"openstack/requirements","branch":"master","topic":"ceilometermiddleware","hashtags":[],"change_id":"I158a8c449d6da6d9e45eba645cf0fd04320a972e","subject":"Add back ceilometermiddleware","status":"MERGED","created":"2020-05-02 22:23:34.000000000","updated":"2020-05-03 07:46:18.000000000","submitted":"2020-05-03 07:44:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725067-1588491896641-23512575","meta_rev_id":"c4ff837365f94c754765e682f71b54118b29cb9d","_number":725067,"virtual_id_number":725067,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-03 07:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-02 23:53:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-03 03:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-03 03:16:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-03 07:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d62de6f36a8e7b75e9a08d43ad59eb9db9087764","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-02 22:23:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8538fd3d59575162ee80fb8524c8016a37ef8136","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 22:39:14.000000000","message":"Patch Set 1:\n\nI\u0027d like to know more about how it wasn\u0027t caught so that we don\u0027t do this again","accounts_in_message":[],"_revision_number":1},{"id":"0a2bcc1f9a082ca9a5b6d418427da4e9c33e4133","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-02 22:47:59.000000000","message":"Patch Set 1:\n\nIt\u0027s not in anyone\u0027s requirements files, not setup.cfg. looks like the only place it\u0027s installed is through a devstack script.\n\nI think for now, short term answer is for us to know not to remove it. Long term, I think the projects that are using it need to explicitly as it to their requirements, our as an extra in setup.cfg.","accounts_in_message":[],"_revision_number":1},{"id":"fb7b3234a7930dd3da6f5a8de1b6e3b92ad0a81f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 23:13:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11e9704b0237b1f29d18b9c28b3beaa58836c4b2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-02 23:42:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bdddf64073ac2dee22613e212842e095c3407b7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-02 23:53:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b6417eeaa043ef7520399a2880da3a7e234add3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 01:24:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc09e35728354178b7f5d0bad48021a1 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404586317234408ea9572de460c09745 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d1e646cda814a8984f05fa2f408b242 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aec6091fa94044bd83fc2686c8dee51a : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/42ef9a1c76f0457babadf29c14a8a8f5 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11d62e1f95ff4f20b154be165b081477 : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1dfd4e0048fc406da0dd084cdb91739c : SUCCESS in 7m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0802688fd7934fc086c86158d24f68ef : SUCCESS in 9m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/21226130a6434d16a2413dc202c0dbe6 : SUCCESS in 4m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/17da31c6a9ed40b7acaa4fefb68c66bd : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/33a6cb2abcae49239108f80f7bae652a : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b025d29b3ce648de93e4bdb94b9509bb : SUCCESS in 14m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5bf70c4dc31e4012a0924ed3cf4a4738 : SUCCESS in 27m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/08c9a25679c24d72a2aa179490d5a375 : SUCCESS in 16m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2e604c439054469ea9e3909e4856ec4c : SUCCESS in 13m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f285e7ab15b4ffbae8e2c2f934d86e3 : FAILURE in 21m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3057a26222d448d896edd88107f3d5a6 : SUCCESS in 4m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ab0f508382ff4a508f2cc70a3ed0b0c0 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5b51b236bd14df4b9bfdbdd98be5d42 : SUCCESS in 6m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/713ff7332d504a3a947a495246759ed8 : SUCCESS in 12m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d05054a3cb1a4c5ca7fc714a621da030 : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bacb6e4b819f457abcb96842623d96d0 : SUCCESS in 51m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8481ee29c4984c2e9bc2d2447b0074af : SUCCESS in 1h 01m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3a8a7087b9d42579b15406921bab138 : SUCCESS in 1h 37m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b32969dbc2eb4a5cb24ae97e09dcbbbc : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7012939ab65363121c80b0263923544ab38c5bf6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-03 01:28:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"79be3026a1badf7370a0fbdaffbe9d647bb7a9f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 03:11:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8d725ad02848cda8f7435c7ce7d2d7 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d18ba46a0d44739af0881bfbc2ce7e2 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94742b4098304456bab94bfb483610e3 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df62f27cbbb64dbba68a13b1348a3f24 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c90212f0c6f34623927f1760d14cb01c : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbe537372c8b44e6b51bf9b98f6b7020 : SUCCESS in 2m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/856d77e3f3614ba7b6c5c12c5dbbff4e : SUCCESS in 6m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b949f429373b4156b9305ce8d9e97101 : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/35cf0750bf8142b9a788468eac551e76 : SUCCESS in 3m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2e66b8d6e49b4b24bc2e9396c7ac16d2 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df8ebf498bdb461cb08e1e8530104ac6 : SUCCESS in 6m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2b5cf08206254cb79212b16d913394c7 : SUCCESS in 11m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eea21ca567734fb9a3260ad6e8bbd0c5 : SUCCESS in 25m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bc20d03505e142f4a590448537f9d42f : SUCCESS in 13m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f8fa36103a9c4f64899f7d369065e99a : SUCCESS in 12m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/203f5fb359d4490c97272c09c33b284f : SUCCESS in 15m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/76dbc6cae341453588038f5bb80c8ea7 : SUCCESS in 3m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/198a44bd723d48798713957109a076a6 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39f97f74a1ca478b85739be680a8035c : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0a65c3c28d274afe859bd4e753ebab6f : SUCCESS in 9m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a3f12bf69b834748bd7f24662918686a : SUCCESS in 9m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fef3dc5e42d14f959238041243c85378 : SUCCESS in 49m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6650e8ab94d4e02b2be0f09c43d9471 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4264916c5c524fea9ce6d2b35a0eb895 : SUCCESS in 1h 22m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a45ac17674ab4b2b963cadf06b4e7468 : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":2},{"id":"7c19dac9d5900215f6eca735524258864ba173e7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-03 03:16:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9a75e8d40be94b040f7fe135ab173cef86bef290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 03:16:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"23a4ff2a44a9150832cc6f3da04f0fedb92c10da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 04:44:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fdfc8c9fa494463bd71d5e5d4a8c7b6 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4c18e847ed4c8f99719747cfa95de0 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/435c35f73e2046c49c9619839abb85e4 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/328d58fa37d24c5a996e9b1d0f9508c7 : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2532881703224db98aab98802b9dbb1e : SUCCESS in 6m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/934adbc4c0da4c08bceb0518579fc44c : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5bcd7e179a0942f8bae3c8483e8f7944 : SUCCESS in 7m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fddef9a26d304a13b705c1a107629d1d : SUCCESS in 8m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0902cfa767604e558056cf9d116ed109 : SUCCESS in 7m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/37ac251ca2f3417bb9fd4d14ca06710e : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56b52dee974849a4953964dbeaf2fdc1 : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c4562689a4c4c9297b606b9e7f35660 : SUCCESS in 30m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7b175c411f04035909aa1005ca7e694 : SUCCESS in 23m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28d33e4f8637408b9206fcb09561b2da : SUCCESS in 12m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7acf03f37d2f4ab9bd2b27e2a4a08e44 : SUCCESS in 15m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f5c74144ac24b1597a0cd87249659c0 : FAILURE in 20m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/51caa234fadb4b4f9e3c4d385ff348e7 : SUCCESS in 4m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/338ede1af7704e209e95063c9a86a56c : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82cd2832ddc94d0c9a0973811e889643 : SUCCESS in 7m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/91796d64dac8485fb0afba7dd9584c02 : SUCCESS in 18m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/df83f7936fa34daebef52bf99c6560bc : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d3ccac16c5a4891b15080648c115880 : SUCCESS in 55m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/477f8b5dd94e483590bf09e8c2514f45 : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1f8d135a841475eb44f4740a3f911a9 : SUCCESS in 1h 27m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fb1daa04dd404950a921bbb374fbb706 : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a0e30c49d3d77681c3d01259a3acc22f84079f62","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-03 04:50:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"36273f700d7efed7c074d690184b880d4d212070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 06:14:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0150f01a1fe143ef82a2472bc771a3e5 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c28071406b6477ea293f0eecdea25a8 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd7746ed8b23410bb692c71446027cd1 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c14a61ac103466da2df7736cadf72d3 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9390fae0a9b2465e811ca96471bf8154 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5c5521db80645da895f627fda79aae0 : SUCCESS in 3m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dbd96efaf20b450f9d56cda2e9f27e60 : SUCCESS in 8m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/50a11c2ff0484fbda774a02c369b1115 : SUCCESS in 8m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c166fea62a734014b1c90c61f252275f : SUCCESS in 4m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/73944bb97f5343bab0446119e89b7a64 : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6437113ceee945469c5ae4d39b117d59 : SUCCESS in 6m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af401e53522a493b81bb65ccca4d6134 : SUCCESS in 28m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/be6edb600e5541f3b84315cdf2fa7f62 : SUCCESS in 24m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1bf161c48835435c8b14b2197b18df44 : SUCCESS in 12m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d8c09aca0e164a26bce582d506f54095 : SUCCESS in 21m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35a7b1ff864948c5b33483c36ae9f0b5 : SUCCESS in 14m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/30b28ff475884a17aaa0c680cd9ffed2 : SUCCESS in 4m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4fce0d5167ff40418eebcaa7c0a962a7 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/44c89c8b3f0f4f5381153b60b4c8dbb1 : SUCCESS in 6m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a040010806eb46df9dbea7fddbd0d6f0 : SUCCESS in 10m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/12e48c43b72e4493a709d54efe0fc6ea : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18248d89ef97429ea7cfddda33cc5360 : SUCCESS in 52m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7df6a9f8b0714ed5a9b9bd16af83d2bc : SUCCESS in 1h 05m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/448b4e5df533483bbbb6b8771419afd1 : SUCCESS in 1h 19m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3e1447cc55b344bb9f9f67f224f3c845 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6020e8018e09b458ebb039de773cf848f0ca80b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 06:14:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2c2759989c729eacc7a6e60a62fe19f8391d2148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 07:44:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b64125270b24d7f8882893077bc0b86 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc39b65ba9a4ea5a46341dfef4aab46 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/395eeba444ab4b8da9efa220fae3ab3b : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4efeb1d7c5694be6a0215200b891e835 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f862d1ea40964b3faf84b8f3357aaaa7 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/173bf681da894e36bcc973a812cc461a : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e95ecaa48b84e3197a1f8f0f0a819d5 : SUCCESS in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85e3d5b6dedf41059f76b3b2e4c1d9a2 : SUCCESS in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/769a8b82e57b40c98af510de10256540 : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/325b971e350841beb596fe9c23394a93 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/baef78776c1341739369b4df45f3b83b : SUCCESS in 9m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1a2f76c46b10499b80b7b9843de1d5ce : SUCCESS in 25m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a263c46a77954c67a736df9bf22e3f76 : SUCCESS in 26m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bbbdeb9d8e5b4b58a8e2963c8944824c : SUCCESS in 14m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c9e3bf7d31b54cd69f6d689389b3c227 : SUCCESS in 21m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04ae61e40c9041ac94575b375ada815a : SUCCESS in 17m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8e7738d8b4049b4960a45d2a4a041b2 : SUCCESS in 6m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2dac5bcd218347d99edcb0ea771e4273 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/562bc2702b484a08b8881c7f0e2c0477 : SUCCESS in 7m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ba1f5de1b39a4511919d7b92edf54511 : SUCCESS in 11m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b50a362f269427f8f122375140bf2d6 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9216e8d9f3634e66ada7fdd2c76a53d9 : SUCCESS in 55m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4933e0acaf3a4473a17e4ea2d01f257e : SUCCESS in 1h 04m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ea2f9fe045945cf99a8a55a4084e86b : SUCCESS in 1h 24m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aae074c9c1ec40a891156d4327688e96 : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b47240df9b02e7bfad6463d9e008f82be913aa2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 07:44:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3f5eb513f63dd0918a8368b4649fb8536358715d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 07:46:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1b9d21ec4b446ea20bca3e72f495ed : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"969cf95aa1dd6b1e62d3e3a08018d8b0f97449ff","revisions":{"c4b7facf7c58232a70f9d83c80ce16d2b5e51452":{"kind":"REWORK","_number":1,"created":"2020-05-02 22:23:34.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/725067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/725067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/725067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/725067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/725067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/725067/1"}}},"commit":{"parents":[{"commit":"75c50a27b0c13c9befc8f8324a31fd08db34df4a","subject":"Remove unused packages from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75c50a27b0c13c9befc8f8324a31fd08db34df4a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-02 22:22:47.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-02 22:22:55.000000000","tz":-300},"subject":"Add back ceilometermiddleware","message":"Add back ceilometermiddleware\n\nRemoved by https://review.opendev.org/#/c/724761/, but needed for\nSahara, even though they do not list it in their requirements.\n\nChange-Id: I158a8c449d6da6d9e45eba645cf0fd04320a972e\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4b7facf7c58232a70f9d83c80ce16d2b5e51452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4b7facf7c58232a70f9d83c80ce16d2b5e51452"}]},"branch":"refs/heads/master"},"969cf95aa1dd6b1e62d3e3a08018d8b0f97449ff":{"kind":"REWORK","_number":2,"created":"2020-05-02 23:42:26.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/725067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/725067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/725067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/725067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/725067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/725067/2"}}},"commit":{"parents":[{"commit":"75c50a27b0c13c9befc8f8324a31fd08db34df4a","subject":"Remove unused packages from requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75c50a27b0c13c9befc8f8324a31fd08db34df4a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-02 22:22:47.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-02 23:42:17.000000000","tz":-300},"subject":"Add back ceilometermiddleware","message":"Add back ceilometermiddleware\n\nRemoved by https://review.opendev.org/#/c/724761/, but needed for\nSahara, even though they do not list it in their requirements.\n\nChange-Id: I158a8c449d6da6d9e45eba645cf0fd04320a972e\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/969cf95aa1dd6b1e62d3e3a08018d8b0f97449ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/969cf95aa1dd6b1e62d3e3a08018d8b0f97449ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
