)]}'
{"id":"openstack%2Frequirements~725191","triplet_id":"openstack%2Frequirements~master~I05a312fb3f2cb44baeecd4b0609f71a21b1c536d","project":"openstack/requirements","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I05a312fb3f2cb44baeecd4b0609f71a21b1c536d","subject":"Switch to new grenade job name","status":"MERGED","created":"2020-05-04 06:44:06.000000000","updated":"2020-05-07 11:21:52.000000000","submitted":"2020-05-06 22:08:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725191-1588802924167-92884968","meta_rev_id":"bf2e7f128ef2959857ec3f5cd6d4adce5af95837","_number":725191,"virtual_id_number":725191,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 22:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-04 17:04:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-06 20:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-04 11:13:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":1,"date":"2020-05-04 18:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-06 20:29:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 11:13:42.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-04 17:04:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-04 18:51:06.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-06 20:29:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-06 22:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96fad67ac1d7235b4cbd82170ef25ce9cc350ac7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 06:44:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c0a0ecf434366f939c87942277563d77bbb5030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:51:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd441f2ea473476381d347687f3b8c3f : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45cf55cb9774440586fb7cc1a43395ea : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5c49c9817bd4f42bb4c9ee803229d96 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/523234a6d53b44aeb6872a47df87a8a4 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2399a3c33201462a8b0856f257f023b1 : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e7a2392fdbc4c50a973ea877a855549 : SUCCESS in 57m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac96c0f979f94716ac7d36896dddc38e : SUCCESS in 1h 01m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d44baaade45a48e6a270f0656272106f : FAILURE in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/b7834dc238794c548975792669ec08e3 : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b1d83e91702a82a5a82687be431fe53ba2774061","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-04 11:13:42.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"596d9db666bc1c86bb86227936fd2cb3b5802a07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 12:58:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48756ac6c3fb45378241bb589dce52d5 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a63e145f5c42c5baf1f8e358ef1c13 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5379aa96dbc74270aa6050fd3e063ce1 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61f9e543d16f4dd5bd06c7b995117e12 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d640011d0cec45078c69d8e34e653fef : SUCCESS in 5m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f7bf411401142e8bbf57f1af76d23cd : SUCCESS in 1h 06m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/44192ea868164973a211e38f5a5a5a5a : SUCCESS in 1h 16m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cf77878269e47b38637b0cfdf26b496 : SUCCESS in 1h 42m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/81ee8ca21b0b461a9ed422392f5383a3 : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3e127df47ade5ecc131da674585ce5444ee23e55","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-04 17:04:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa4e52d793de5bb58009ed9f94fe8c8b73f2f48d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 18:51:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4a71de5dbd0954cb309f540c823397c42e752b3e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 20:29:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c81866216b45520add01c3401ab1662911f6fbc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 20:29:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"453ccc525862b563bb63155dd85b7dd86aed8d33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:08:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e345ce6030476da677f995b44a9da3 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c7b2067baa4bbdbbe855ce6d50030a : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8e7235dec8e4dacb342d5085c7af552 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55aba75e42f243f2bad79f2e9de49c25 : SUCCESS in 8m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3bb50d158c047f8aaa5a9a921d8e2dc : SUCCESS in 6m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01063784291d4463964074eaaeef7942 : SUCCESS in 1h 01m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eba5c26a383041188cf8e6ebbb5e2545 : SUCCESS in 1h 36m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1bcbd4f82810473ea44356bde7048c82 : SUCCESS in 1h 37m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/0e60e4e99904423ea754deae7bf44758 : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5603fffe17124cf43da95aeed60f5ee513993bbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:08:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"490d9564f123c7bf8ccdf85b43b14e22a194ed63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:12:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6933f20d6d438ca6a57abc5ffbeac4 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c0ce2135f0e088c52ce764394d5728bd479e41f3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-07 11:21:52.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 7d3a23d79bea8a66890751f1c123db6e04514e72","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6ba3f074bd49d4d505a96dced907c756b1806eb","revisions":{"c6ba3f074bd49d4d505a96dced907c756b1806eb":{"kind":"REWORK","_number":1,"created":"2020-05-04 06:44:06.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/91/725191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/725191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/725191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/725191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/725191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/725191/1"}}},"commit":{"parents":[{"commit":"a5e2d99188a8ce9f10b4c70cb74b070db589fb0a","subject":"Merge \"Add back ceilometermiddleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5e2d99188a8ce9f10b4c70cb74b070db589fb0a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:43:10.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:43:12.000000000","tz":120},"subject":"Switch to new grenade job name","message":"Switch to new grenade job name\n\nThe integrated gate templates has switched to the new grenade name\n(grenade-py3 -\u003e grenade).\n\nRename the job here as well so that grenade-py3 is not used on master\nand stable/ussuri anymore.\n\nDetails:\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014602.html\n\nChange-Id: I05a312fb3f2cb44baeecd4b0609f71a21b1c536d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6ba3f074bd49d4d505a96dced907c756b1806eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6ba3f074bd49d4d505a96dced907c756b1806eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
