)]}'
{"id":"openstack%2Frequirements~725488","triplet_id":"openstack%2Frequirements~master~Iec6b639fce37135f8840240b5d2a1a168bf9bb4e","project":"openstack/requirements","branch":"master","topic":"validations/global/requirements","hashtags":[],"change_id":"Iec6b639fce37135f8840240b5d2a1a168bf9bb4e","subject":"Add validations-libs to the global-requirements","status":"MERGED","created":"2020-05-05 07:17:59.000000000","updated":"2020-05-06 19:40:44.000000000","submitted":"2020-05-06 19:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725488-1588793830699-57c03af8","meta_rev_id":"6d0991f06bef5c70feeff1a6ca0bbe2979517981","_number":725488,"virtual_id_number":725488,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 19:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-06 12:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-06 15:47:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-06 15:47:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 12:30:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-06 15:47:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-06 19:37:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08b58ff17be50a6b8f077a52ae27b5eb96636939","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-05 07:17:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8c509b16f51f1bc958fc63ac22b00d450b42d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 09:30:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465b5744f2ed4217b27079a865408950 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1874c5048b114489a06a4532c57aab9f : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b51003b92a5436385d76c682226a5bb : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0768a73ebc8a42d5b691a67339583c1c : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa2b77266fd54349b4ce11d2524ef78e : FAILURE in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ad7ec0ce70c4cbfb892732ae275c90c : SUCCESS in 59m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2427476c3bee4547a1770c1b2adae3e4 : SUCCESS in 1h 50m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/163e8f28b37d4378aa637f16ec134fda : SUCCESS in 1h 23m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9a30e6b8be9e459382944b1afd7315ed : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d08bc0f16dfb1b586939af47f1101cc6968c05d0","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-05 11:41:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fac203ce041d16a196762b549c72ba719aeeccf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 13:01:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25b055d44624ae6a89488a07638f1b0 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b206f7dd2a64506a39c14ba21458329 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc3403a64b2d41cc9f3576a911893849 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f90d01d8a230441ca4b1e48a23c9f3f7 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/356363fe897f4a9a8d397873fd8543d0 : FAILURE in 7m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c56c826b5da4de38d14045719f1ca04 : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3fa1664bd5134d4a860d5f9b78951586 : SUCCESS in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb4a267855b640a0867963225955e710 : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/378c158b55b444f3818c51f7c2c8d485 : SUCCESS in 7m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ac913fcdcba4408d8547c3e4dba92504 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/faba15f9a4864cefa6aa04001cd74c0c : SUCCESS in 12m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a5c3e28dff3a42ebabf9887cfbf9a81b : SUCCESS in 25m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1b259bcd56ba43a8b1ebd71d343fdb0e : SUCCESS in 29m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1629df18d9d34915b2abb377886e2ada : SUCCESS in 17m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cd2e686b906d443ab2642b79162bc895 : SUCCESS in 22m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8bc5dd4a98ac41129d5fa65274b9a2ec : SUCCESS in 24m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cd2b1e11f1a74b5b95b5429660d0d945 : SUCCESS in 6m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/adc23c3246a948d39d8fffbe049c397d : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d73a494a9a0c482c8c939a26d3cefa7d : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/860fae40ddad41ee947bc85f1ba42360 : POST_FAILURE in 15m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/09052dd0ebe8496684fa2ce6913d08fb : SUCCESS in 16m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5eb725b92d6d4ae3af347cc4211d300a : SUCCESS in 55m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b6862c70135e401eba150de770641e6b : SUCCESS in 1h 16m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cbc40828f364d5f9a6b3ef14caf1207 : SUCCESS in 1h 16m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/091812b80b154e17a0ad02432d55f8fd : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":2},{"id":"390d41a105b4d967f5988b0c77871588277da413","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-05 13:39:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"122d648683fcf388f44d9ca6bd754b1144780b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 15:19:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a7930dd6f54422b457417b04d81f10 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abb8794505b49b18e93d777ac223e7d : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6374bd9b19724ba6bba3a7fbd09925dc : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aff78be793bf44908b64beb5bba78b2b : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0482eca3c2324ae9bdf4e9d14e7ba2d0 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/473aa845ea2e498f9f03d528771cff3f : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b110202f40f14d79af585a71e2a2d694 : SUCCESS in 13m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/94108fcf11bd4df7aa78c208f38a1961 : SUCCESS in 15m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab9373a07f634c45bb02935ff9b59a1e : SUCCESS in 6m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2c43451539a645dbad89dd4ce1bfac90 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d99dd1b11e004fb993a0a0c82250fdd8 : SUCCESS in 11m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/071c2ab458084a9bbd6ffce5ea824eb0 : SUCCESS in 18m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3cd0c6124df14ef6831fe613e4dc64cc : SUCCESS in 27m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d29025ca0c37404195fa86342ecf075b : SUCCESS in 17m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/29b3d07a4bd94167879b97e5a27adb3a : SUCCESS in 16m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c564162d20724c9aae86240666520b8f : SUCCESS in 17m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/46f39c9804f94c78af00a28bc65160a4 : SUCCESS in 11m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/680b9a970649489985d72d76aa5808ec : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8658660dd5c4654ae9d1767b1647bd2 : SUCCESS in 9m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1d859bce12484e9cb88da498621d9799 : SUCCESS in 18m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e019e1521cc04295868ed441c300473e : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0c0ef4ecc7f4ff7b6364b1615a8a2dd : SUCCESS in 53m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/60abd612fd95489cbd7757a3a4130179 : SUCCESS in 1h 16m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5cdd620a666449eb7087c4aad5f4350 : SUCCESS in 1h 29m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0e29dcc705374297917b0a73b0149a82 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":3},{"id":"7af9a34be0ba5f9933de8f4ea5d473e27bada107","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-05 15:21:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"01b6b03c1e0078fa688a659e99ca649e2bac4bf2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-05 15:45:16.000000000","message":"Patch Set 3:\n\nI think we still need the questions answered in the commit message as well\n\nhttps://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements","accounts_in_message":[],"_revision_number":3},{"id":"4e38f4ee6ba2ceccf582cd3404f341b0ef4c65e1","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-06 08:30:20.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"2acd5e619482e1224f80720090c60c1954a20dfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 09:51:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9797503bcfef472d8818b7c66d2c620e : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964c4a797b9f4c6b87a6eb3527fbd781 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aba2d8b81244956bad77da51372c40f : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09145e8856244527aef6891125e8005c : SUCCESS in 8m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e37a79bb015443fd92a77896bf652342 : SUCCESS in 7m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e86441cf57cd4804a52d5f227b0ac5d5 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a3a6a67b04fa4588a8c4ffec1f0c4fc9 : SUCCESS in 9m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6f71ecc35dbf434d9aa907e6d40d568f : SUCCESS in 13m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5df376e347514f23a5e60e8a22a30e3f : SUCCESS in 7m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ccc010f9c9934d789ac95d117bacddf0 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/33b899a7505a4e1a87dd088a75cd8e2c : SUCCESS in 7m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a357a2ae76b04b50a2b624d25bd025c1 : TIMED_OUT in 1h 02m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/459aeb9f68504cb68780f2aa93e29e4a : SUCCESS in 24m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0537f6f8c654472084524dc6560d7f0f : SUCCESS in 16m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1a21aa1f267d4af0afc2023e6ccc29b4 : SUCCESS in 21m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40dea4fa5df8447f81853fbf602b719e : SUCCESS in 17m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/054124101dac4f3a860524da1d71ed74 : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4b982182740e47fb9315aa88b79a95e1 : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1624b160ae3945dfa66c42878b468133 : SUCCESS in 8m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f4526d8e92464634887b37b8e4710b2c : SUCCESS in 13m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef56ced292594270b3ef2b572cdd755d : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ba11b4b3c0d4b29b2246d34dde07acc : SUCCESS in 52m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7f659b188a74ca68433836ede377271 : SUCCESS in 1h 03m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20ace9728bbd456d9c523d9e3226ea64 : SUCCESS in 1h 18m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ee6fa61979c14099a0dfb17e2bcd9eaa : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":4},{"id":"7a1553d421d0250dd50190607ef22f301a7d70b2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 11:56:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a8406f880e00e515e06c5463746ea8cef4fe20d9","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-06 12:19:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3f60e37d4d3ccd9155df92489a71d8becba34b01","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 12:30:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cbc5f931dfa5663e9ceed65df8909083927f91b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:51:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc5068281ba4bb5904c7fbc6da605b4 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4656c00d0e441b1bdd992bcbf755cdd : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d391474650164d2cb701fc3cba5b2192 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b21e50ca59654239a5cec73ab81be4fb : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4589ca29cf8c4468a826c92ddc4b34da : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/693fa5076a4c418784214e26eaf844dc : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b4e8b57baf64180ab4d7b5a59e47f60 : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7bb1b746a8ef4c02ae687b91edf309e1 : SUCCESS in 10m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c5d50d8c427a43d496d5ae01847d832e : SUCCESS in 5m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8aac30158ed24579ae9510b780910844 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e21a90ee28f14a6898d31a6c8784056f : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dcdf5e42fd064624b318b22d0ee50af2 : SUCCESS in 18m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0948da995bab4f5680f1a6069cf6f5ce : SUCCESS in 31m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d9e635a66ccf4a2ca3939333b0a7e97c : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/12bb9fa7e3e74e41a3353009ff67753d : SUCCESS in 18m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c06f9c869ea0428da35733cf666df1ac : SUCCESS in 17m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/68dd773a3e88460eacdc38316a852395 : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0cfd7d6b19934cb68b1619109e27b623 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f23b8a5c97c4780910974386ba853fe : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/58ed0c2d2bdd4146882384a10cc11d41 : SUCCESS in 11m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a77f1301222b4986a0e9a51a60f8011e : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7791b626f53845bda8f0ebfdfe7f598f : SUCCESS in 51m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbb94fcad76c49df8815c2b2b2d9dfff : SUCCESS in 1h 11m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/792228ce88fb4d6b8963159773c9bd1c : SUCCESS in 1h 25m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/85f512aff1384f1db1fb7ce4d735d642 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":5},{"id":"e97165070359f414535caa41bc7e9b4e15008f03","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-06 15:47:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"afa2a12617b6abdced50e9bbf37093fe2589e314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 15:48:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8d0f8b512518712bde9cf18a782bf9fcfc6e7e3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 19:37:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32e542a13cd403db85d1b5e6702acfa : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab637bc740814989ba2513a1e7dae4da : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1010fb9cabc4fa8b6695f46092e3314 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50e2753104644977a3d2b5fea103ee74 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49bdd6e3f27341288f81ecf7aed63add : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91d0625fbffe42019f7b2bc7d84f62be : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d12d49664ee450b87deaf7763b4d233 : SUCCESS in 12m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/231383e842534cb7995ecb8477533544 : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7b19313f05304e5a9e1072582b3ee5bd : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2efaeb5b59f744e6935b541d167342a8 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0a9388bf260401492245847d618904f : SUCCESS in 7m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/24d205d37499481494bc3ddcb3f7b477 : SUCCESS in 17m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/486d7ff34548429eba0ccf0626df51e0 : SUCCESS in 33m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fe520b5fbd5546838badf781e6d8ec79 : SUCCESS in 16m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b656f89320f1456d9199b331255e5e8f : SUCCESS in 14m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e0f5eef6e0848c9b5fa67a5766751e4 : SUCCESS in 23m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/efe1b4bbf28c44ada1f53929a3f026a0 : SUCCESS in 5m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e14dd92cdc7f4a64898993242b00e467 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8acad4095bf84af0ad8e10d7489444ae : SUCCESS in 8m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/36e9bc4d47d748cbb1fd6aa148419fdc : SUCCESS in 11m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/def6e8f599e0497e8bfb7a5defd95e17 : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8603e46e58e4def9398dcc13356a1f3 : SUCCESS in 55m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5423e506ea7f47abae3a3cc2603c2a11 : SUCCESS in 1h 04m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8682ad2543149149a291c4dcf0f5f6b : SUCCESS in 2h 00m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c886e1770f424b3a9c3b480a843d2280 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":5},{"id":"c4cd051caac58c854efea9a460f05b68617c9bff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 19:37:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"e4537bc6f22cd54a97aa9d70a12d7acb4d205ebf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 19:40:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1484654bdfb9431c961711959d12d270 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5368475655b129ed71a6c207220570a36a93b3de","revisions":{"1317a5608acc09c18731a39aea5cd5619a25965c":{"kind":"REWORK","_number":1,"created":"2020-05-05 07:17:59.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/88/725488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/725488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/725488/1"}}},"commit":{"parents":[{"commit":"440a1c8409b32d8cf6478e2f6da17b23185fb606","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/440a1c8409b32d8cf6478e2f6da17b23185fb606"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 07:16:47.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 07:16:47.000000000","tz":120},"subject":"Add validations-libs to the global-requirements","message":"Add validations-libs to the global-requirements\n\ntripleoclient needs the validations-libs as dependencie, we need to\nadd the validations-libs to the global-requirements:\nhttps://review.opendev.org/#/c/714629/\n\nChange-Id: Iec6b639fce37135f8840240b5d2a1a168bf9bb4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1317a5608acc09c18731a39aea5cd5619a25965c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1317a5608acc09c18731a39aea5cd5619a25965c"}]},"branch":"refs/heads/master"},"f3e4dc9c1c2f1bb5975c67647062c206a12e1bc8":{"kind":"REWORK","_number":2,"created":"2020-05-05 11:41:01.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/88/725488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/725488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/725488/2"}}},"commit":{"parents":[{"commit":"440a1c8409b32d8cf6478e2f6da17b23185fb606","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/440a1c8409b32d8cf6478e2f6da17b23185fb606"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 07:16:47.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 11:40:48.000000000","tz":120},"subject":"Add validations-libs to the global-requirements","message":"Add validations-libs to the global-requirements\n\ntripleoclient needs the validations-libs as dependencie, we need to\nadd the validations-libs to the global-requirements:\nhttps://review.opendev.org/#/c/714629/\n\nChange-Id: Iec6b639fce37135f8840240b5d2a1a168bf9bb4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3e4dc9c1c2f1bb5975c67647062c206a12e1bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3e4dc9c1c2f1bb5975c67647062c206a12e1bc8"}]},"branch":"refs/heads/master"},"1ef6e1acc3ee4aabe8ec714b56eb412036746092":{"kind":"REWORK","_number":3,"created":"2020-05-05 13:39:36.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/88/725488/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/725488/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/725488/3"}}},"commit":{"parents":[{"commit":"440a1c8409b32d8cf6478e2f6da17b23185fb606","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/440a1c8409b32d8cf6478e2f6da17b23185fb606"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 07:16:47.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 13:39:24.000000000","tz":120},"subject":"Add validations-libs to the global-requirements","message":"Add validations-libs to the global-requirements\n\ntripleoclient needs the validations-libs as dependencie, we need to\nadd the validations-libs to the global-requirements:\nhttps://review.opendev.org/#/c/714629/\n\nChange-Id: Iec6b639fce37135f8840240b5d2a1a168bf9bb4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ef6e1acc3ee4aabe8ec714b56eb412036746092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ef6e1acc3ee4aabe8ec714b56eb412036746092"}]},"branch":"refs/heads/master"},"29149a9357bf4f8a69fdc37a0ea924a1da922204":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-06 08:30:20.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/88/725488/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/725488/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/725488/4"}}},"commit":{"parents":[{"commit":"440a1c8409b32d8cf6478e2f6da17b23185fb606","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/440a1c8409b32d8cf6478e2f6da17b23185fb606"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 07:16:47.000000000","tz":120},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-05-06 08:30:15.000000000","tz":0},"subject":"Add validations-libs to the global-requirements","message":"Add validations-libs to the global-requirements\n\ntripleoclient needs the validations-libs as dependencie, we need to\nadd the validations-libs to the global-requirements:\nhttps://review.opendev.org/#/c/714629/\n\nAnswers to the basic requirements questions.\n\nIs the library actively maintained?\n\nYes, the libraries is actively maintained and mainly used by TripleO projects\n\nIs the library good code?\n\nYes the libraries is covered by CI and Unit tests\n\nIs the library python 3 compatible?\n\nYes\n\nIs the library license compatible?\n\nYes\n\nIs the library already packaged in the distros we target (Ubuntu latest / Fedora latest)?\n\nYes\n\nIs the function of this library already covered by other libraries in global-requirements.txt?\n\nNo\n\nIs the library required for OpenStack project or related dev or infrastructure\nsetup? (Answer to this should be Yes, of course) Which?\n\nYes, it\u0027s required for python-tripleoclient and TripleO projects\n\nIf the library release is managed by the Openstack release process does it use\nthe cycle-with-intermediary release type?\n\nYes\n\nDo I need to update anything else?\n\nNo.\n\nChange-Id: Iec6b639fce37135f8840240b5d2a1a168bf9bb4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29149a9357bf4f8a69fdc37a0ea924a1da922204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29149a9357bf4f8a69fdc37a0ea924a1da922204"}]},"branch":"refs/heads/master"},"5368475655b129ed71a6c207220570a36a93b3de":{"kind":"REWORK","_number":5,"created":"2020-05-06 12:19:25.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/88/725488/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/725488/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/725488/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/725488/5"}}},"commit":{"parents":[{"commit":"440a1c8409b32d8cf6478e2f6da17b23185fb606","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/440a1c8409b32d8cf6478e2f6da17b23185fb606"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-05 07:16:47.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-06 12:19:13.000000000","tz":120},"subject":"Add validations-libs to the global-requirements","message":"Add validations-libs to the global-requirements\n\ntripleoclient needs the validations-libs as dependencie, we need to\nadd the validations-libs to the global-requirements:\nhttps://review.opendev.org/#/c/714629/\n\nAnswers to the basic requirements questions.\n\nIs the library actively maintained?\n\nYes, the libraries is actively maintained and mainly used by TripleO projects\n\nIs the library good code?\n\nYes the libraries is covered by CI and Unit tests\n\nIs the library python 3 compatible?\n\nYes\n\nIs the library license compatible?\n\nYes\n\nIs the library already packaged in the distros we target (Ubuntu latest / Fedora latest)?\n\nYes\n\nIs the function of this library already covered by other libraries in global-requirements.txt?\n\nNo\n\nIs the library required for OpenStack project or related dev or infrastructure\nsetup? (Answer to this should be Yes, of course) Which?\n\nYes, it\u0027s required for python-tripleoclient and TripleO projects\n\nIf the library release is managed by the Openstack release process does it use\nthe cycle-with-intermediary release type?\n\nYes\n\nDo I need to update anything else?\n\nNo.\n\nChange-Id: Iec6b639fce37135f8840240b5d2a1a168bf9bb4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5368475655b129ed71a6c207220570a36a93b3de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5368475655b129ed71a6c207220570a36a93b3de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
